)]}'
{"id":"openstack%2Fsahara-plugin-cdh~686732","triplet_id":"openstack%2Fsahara-plugin-cdh~stable%2Ftrain~I5de5a2109f0b50adc209008738920fb11fd0fd76","project":"openstack/sahara-plugin-cdh","branch":"stable/train","hashtags":[],"change_id":"I5de5a2109f0b50adc209008738920fb11fd0fd76","subject":"Fixes on CDH for python 3 compatibility","status":"MERGED","created":"2019-10-04 14:19:44.000000000","updated":"2019-10-09 16:10:13.000000000","submitted":"2019-10-09 16:06:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686732-1570637219090-0f5a3271","meta_rev_id":"ec99a4c1fed6a8e2a43700a3b9eaebacea53eb25","_number":686732,"virtual_id_number":686732,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"date":"2019-10-08 15:13:17.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 16:06:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-09 15:54:58.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-10-08 15:07:16.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-10-09 15:54:58.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-10-08 15:12:48.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 15:07:16.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-10-08 15:12:48.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-10-09 15:54:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-09 16:06:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58c3b3a9375416b06a5f3c1b28e2a397dc07d102","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-10-04 14:19:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e76d3773d57c9bb1a39a1ee0d26c7a7f12fec634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 15:23:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f3c7db31d7a433bab5c55a5670c9db6 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cfa2accd65d40cfae378b3c9a17cb0f : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b5b4f1bb50247beacb2cd5ff402cd36 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2e3e62fd0664c8192eab6a84b8e5305 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86bd9614b0bb4657b172cce708e94430 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b8a7456df949c0862b2813deb31128 : SUCCESS in 6m 48s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/1fa32f10b04e44f099d9191291b0633f : SUCCESS in 1h 02m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17d7ab606f1d1f5a3146f7acae8227dc8ccc7d52","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-10-08 15:07:16.000000000","message":"Patch Set 1: Code-Review+2\n\nnot sure if we still are supposed to involve tony?","accounts_in_message":[],"_revision_number":1},{"id":"0e3acb71ab7aed3bc6d8b7300001ac3f78a10828","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-10-08 15:13:17.000000000","message":"Patch Set 1:\n\n\u003e not sure if we still are supposed to involve tony?\n\nGood question: Tony, are we approved as part of the stable team? Especially Jeremy as PTL :)","accounts_in_message":[],"_revision_number":1},{"id":"14374b8d64da243ed878ba290426f777a4a01217","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-09 15:54:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nIf you are in stable core you should be able to approve this.\n\nLooks like this is merged to master and an important fix to include in train, so you can blame any further confusion on me. :)","accounts_in_message":[],"_revision_number":1},{"id":"e7f0a26355f8aa6ce1bd17a7dd84de5c9855539d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 15:55:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6cdbbb655f7d7586afb0ffdefbdc54ce0003ffc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:06:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7239e25f0c84a948a3eb7f92eef3829 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6eaf1f1082941e59ff7c2c78aeb82a5 : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b073febcb224234ba40e30127e8e13f : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d1bf80bf6a64aa7882f13b143a534a3 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aeb3a3ad248449c98862aa7a196f5d68 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/363f527d3b58422fb636b352c1d963b4 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"712480f6af972e6cfa9ab045ec7990173693bf21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:06:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec99a4c1fed6a8e2a43700a3b9eaebacea53eb25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:10:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c61dd384734fd69c4a7c581423b876 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4c8cbbcada44b448a7d4bb074a46e4c792482d7","revisions":{"c4c8cbbcada44b448a7d4bb074a46e4c792482d7":{"kind":"REWORK","_number":1,"created":"2019-10-04 14:19:44.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/32/686732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/32/686732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/32/686732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/32/686732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/32/686732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/32/686732/1"}}},"commit":{"parents":[{"commit":"08aa9731dd1cf94690f13b4e1124ead5bc5e197f","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/08aa9731dd1cf94690f13b4e1124ead5bc5e197f"}]}],"author":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-07-31 18:17:33.000000000","tz":-180},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-10-04 14:19:31.000000000","tz":120},"subject":"Fixes on CDH for python 3 compatibility","message":"Fixes on CDH for python 3 compatibility\n\nStory: #2005914\nTask: #34174\nChange-Id: I5de5a2109f0b50adc209008738920fb11fd0fd76\n(cherry picked from commit 9d443815fcfdd3fe11b6cbe3c47660920a2a8fc5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c4c8cbbcada44b448a7d4bb074a46e4c792482d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c4c8cbbcada44b448a7d4bb074a46e4c792482d7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
