)]}'
{"id":"openstack%2Fsahara-plugin-cdh~687649","triplet_id":"openstack%2Fsahara-plugin-cdh~stable%2Fstein~I5de5a2109f0b50adc209008738920fb11fd0fd76","project":"openstack/sahara-plugin-cdh","branch":"stable/stein","hashtags":[],"change_id":"I5de5a2109f0b50adc209008738920fb11fd0fd76","subject":"Fixes on CDH for python 3 compatibility","status":"MERGED","created":"2019-10-09 18:35:37.000000000","updated":"2019-10-14 18:08:46.000000000","submitted":"2019-10-14 17:47:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687649-1571075220726-d762e54f","meta_rev_id":"489b6a6b7220330899312d7791fd2602cd97a1c7","_number":687649,"virtual_id_number":687649,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 17:47:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2019-10-14 16:24:29.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":1,"date":"2019-10-14 16:24:29.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 16:24:29.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-10-14 17:47:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56cc00bb26de8d148e64a117fa29f3329902adb7","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-10-09 18:35:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f16b240b88b384454b0d9ca90d02f79eb88d970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 18:44:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ae8268a806449b59d9930b0b878def0 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afca2c0d2e254e40bbbf621f0f001b57 : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3d0bb3351434fd180f237bfde4ae1bd : SUCCESS in 4m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e314d4574d7a4b018ce941fc6d27cab7 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4ab19b9b10344e5ad9dc690836f6e00 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a452248dc2734f3898b0f48cb3c988ab : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/111ab557a21249a9b977d014f2dbe636 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d1343ba8f44703ea160b64cba3c8874fb64f06a4","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-10-14 16:24:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf3f98a36a0232e389ab48b4497d7d9d2b43e1b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 16:24:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ccdf4ac21ba6518a4d2b50da1a357f028ad18b7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:47:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdfad0420eba4f33859e167ca68be36b : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/114ebc957e3e432da047b095092844d6 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27c0401bf9164080a585ebeba9afd41f : SUCCESS in 3m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e1dada6cdbfe43ca94d8e1a030b08c2e : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24b1282098f54c2d9fd22704875945d9 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8fd480578f441fc80bc36cffd5af226 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d9ce27b3974311a364cda0393b7e60 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"eadcbb79db92dfbc0dd4246e8056831fdfb40b51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:47:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"489b6a6b7220330899312d7791fd2602cd97a1c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 18:08:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c15f735ff145298f7ec35944a48ea4 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c4b42448bb3b341e150bad64e4e2d3840902d7b","revisions":{"3c4b42448bb3b341e150bad64e4e2d3840902d7b":{"kind":"REWORK","_number":1,"created":"2019-10-09 18:35:37.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/49/687649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/49/687649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/687649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/687649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/687649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/49/687649/1"}}},"commit":{"parents":[{"commit":"b73a04d0c8c14686e2ade370406609277032bcb7","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/b73a04d0c8c14686e2ade370406609277032bcb7"}]}],"author":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-07-31 18:17:33.000000000","tz":-180},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-10-09 18:35:08.000000000","tz":120},"subject":"Fixes on CDH for python 3 compatibility","message":"Fixes on CDH for python 3 compatibility\n\nStory: #2005914\nTask: #34174\nChange-Id: I5de5a2109f0b50adc209008738920fb11fd0fd76\n(cherry picked from commit 9d443815fcfdd3fe11b6cbe3c47660920a2a8fc5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/3c4b42448bb3b341e150bad64e4e2d3840902d7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/3c4b42448bb3b341e150bad64e4e2d3840902d7b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
