)]}'
{"id":"openstack%2Fsahara-plugin-cdh~756544","triplet_id":"openstack%2Fsahara-plugin-cdh~stable%2Fvictoria~I546b8a7021ccc172a7c6a711f63d2ddc73da5d42","project":"openstack/sahara-plugin-cdh","branch":"stable/victoria","topic":"fix-requirements","hashtags":[],"change_id":"I546b8a7021ccc172a7c6a711f63d2ddc73da5d42","subject":"Fix reqs (focal)","status":"MERGED","created":"2020-10-07 15:58:33.000000000","updated":"2020-10-07 16:50:45.000000000","submitted":"2020-10-07 16:49:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756544-1602089344077-3e89e6fb","meta_rev_id":"b9b261db8bd123064e87b3a8bcd532951a0a16a1","_number":756544,"virtual_id_number":756544,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-07 16:49:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":2,"date":"2020-10-07 16:27:14.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":1,"date":"2020-10-07 16:34:25.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-07 16:27:14.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2020-10-07 16:34:25.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2020-10-07 16:49:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b53e807c695c2831547544c355e7af85f8c7c48","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-10-07 15:58:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b217b360881604c03809fc33c7b97d3b982627c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 16:17:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a7747678bee64cf69f465db7e9e7abde : SUCCESS in 2m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19d3a049037d446e91e5713689c2c631 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f057ea968cec4dd894c66966de188da3 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c74750630b344208ad605ff1a8d6a5c9 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0b9e728bc3748099a43056b07fa3aa9 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24e2bde92c43445099aa2693135eb85e : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97ee9ee5f33b46ffaa0dc68dcab8ab91 : SUCCESS in 3m 52s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/3471a7ec28c0408fa763a3752d426b47 : FAILURE in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3daf428f8e89762adc3e63183d0e1ab5b8f21be7","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2020-10-07 16:27:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d42f5c9cb3b3cfeae05d665263a39fb9b1cbd3ae","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2020-10-07 16:34:25.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"178683b3e3cec3b531e3913c0896a9ebaa6db9c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 16:34:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3b2d30b130b8d277c20e3770ef55c5fb73f7b97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 16:49:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fd17a354009d45cdb79e88807c5b1eb2 : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ab2e1c850a44a099a8eff7bcc3f0023 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65b95a3219e1450a90ea240c34751db6 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83e150bf162c486694d5a7fddceb898d : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd32a523cd50409193f4d10a9ebfb8fb : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc0b7388664b4a6f9b5f581238ed0f35 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e252370780a4437489fb88a2899c713a : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8fccdf5816b463c17bf0137d611b0cc52786f7eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 16:49:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7fb78102ae96d6904fbfef435b40303571f81d9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 16:50:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad48ae076abb4e3b911167124cdb01fd : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/468c66ee84494ef89b96576c7aa9a8c7 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32653fcc72893b1a36744a23e2bdc37aecee159d","revisions":{"32653fcc72893b1a36744a23e2bdc37aecee159d":{"kind":"REWORK","_number":1,"created":"2020-10-07 15:58:33.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/44/756544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/44/756544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/44/756544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/44/756544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/44/756544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/44/756544/1"}}},"commit":{"parents":[{"commit":"8429c530fad71e04e3bfb9bf26d21fbe9a8bc435","subject":"drop mock from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/8429c530fad71e04e3bfb9bf26d21fbe9a8bc435"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-10-06 19:44:10.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-10-07 15:57:53.000000000","tz":120},"subject":"Fix reqs (focal)","message":"Fix reqs (focal)\n\nApply the fixes already applied to other repositories\nto address compatibility issues with the newer test\nplatform and for cleanup reasons.\n\nChange-Id: I546b8a7021ccc172a7c6a711f63d2ddc73da5d42\n(cherry picked from commit 5f49cb668918214aa5d6c8298f5946923ba7db5d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/32653fcc72893b1a36744a23e2bdc37aecee159d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/32653fcc72893b1a36744a23e2bdc37aecee159d"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
