)]}'
{"id":"openstack%2Fsahara-plugin-cdh~822466","triplet_id":"openstack%2Fsahara-plugin-cdh~master~I7805ba2ff67335d23917de8b04d6aeaeff83120f","project":"openstack/sahara-plugin-cdh","branch":"master","attention_set":{},"removed_from_attention_set":{"34375":{"account":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"last_update":"2022-06-30 11:16:46.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-06-24 02:52:29.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}},"hashtags":[],"change_id":"I7805ba2ff67335d23917de8b04d6aeaeff83120f","subject":"Remove py36 and py37","status":"MERGED","created":"2021-12-21 10:24:54.000000000","updated":"2022-06-30 11:21:50.000000000","submitted":"2022-06-30 11:16:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822466","meta_rev_id":"94d611d03fd3413e41bcfcff6ef0914fe1a2cfb9","_number":822466,"virtual_id_number":822466,"owner":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-30 11:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-30 00:51:53.000000000","_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"value":1,"date":"2022-06-30 02:57:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-30 00:51:53.000000000","_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-21 10:37:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-27 07:41:08.000000000","updated_by":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"reviewer":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"state":"REVIEWER"},{"updated":"2021-12-27 09:38:19.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-06-30 02:57:12.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"dbf876a8d0eeccd2e50e014f9baceb9cb380261b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2021-12-21 10:24:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0089a10f8143280796a4249ec50e437c3b6746a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 10:37:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1ebc8eb7fa6f44c3a8f8e83846b9b07e : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2dcfdde35dd4f9eb1877d13fae4051a : FAILURE in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cbbda6c1c5642f49727c7977bd18a28 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3dedccb6b0040ca80a8748e82dd8728 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50a309e069b740d5a776d86725203dc6 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/206cefc80ffb44e8a03898f9047c5355 : SUCCESS in 6m 17s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/ea0a7d4422bd4b88b317a891a2c24de3 : FAILURE in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9445e001696ac655f742c4d5bddf4539a9d085c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2021-12-24 01:41:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cffe8eb518fc5504218ae9df598d14e23f60c135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 01:50:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d58525b1b2844587a691060075795bf4 : SUCCESS in 2m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/92e61fd86b5d4a099a544d97a304b9d4 : SUCCESS in 2m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cc75eed9caa4f10a5901e37a493f92a : FAILURE in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47cbde6858fb4a9890dbdacb45dfb32c : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06c882ef5ce4459daf4cf63566b6d615 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05e54d015f4a49ed9b049c730559348d : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc8890abf50a40ea99222a39718aaa63 : SUCCESS in 6m 43s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/713e50a1473e4d7285e0ffd888cd3e77 : FAILURE in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"266d0f2c00014e9811342edcdccd1ef95b6dfc8f","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-12-27 07:41:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e256783972e86684572b6faef4cd680ac1243c92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2021-12-27 07:44:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b3c435b6f8d7b19b4d420b1354396f21d9197bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 07:57:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dd66fd5f0ab64c57be32d220e93759d6 : SUCCESS in 3m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4e3f179472c240e0bf1aeea37c9ec081 : FAILURE in 2m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9c313fd1d374aa68ea1ec63f313b6fa : FAILURE in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b21c1da0105467d9c8d7580212abd06 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f5fc04c4e32403ab378150f81ac30fb : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b248a0a2a5b74038b4e19bec0c2e3162 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a7052fec594128b706313a4f4916d0 : SUCCESS in 5m 56s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/db8143574f1746e5b11d9047dbf50b7c : FAILURE in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d1485653c15c4ce728bee1f97b7eb18653737d98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-06-17 07:11:38.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e9de561c2a6f4bb6de53e663c3aa635ee0d325e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 07:28:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bc1f47e6abac4f61a73ffa6e972c5a8c : SUCCESS in 2m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca24aa6c1cf64ae0a6e740fecb17a8d7 : FAILURE in 3m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4feb3cd1b69a4669a8afbf5093feafe7 : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5f8505971c4240a2fcfca6b1f6ea30 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/897d03f064f54efa9e7afa8ac8e0df0a : FAILURE in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f921d2bd57c4c04b8fe47d26c7a5146 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebba869e0a8e4a78b331596164870d44 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a70049157b44dc807552ae05210d79 : SUCCESS in 6m 16s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/e7bf8f1029fe4e75941c6932aa4d8310 : FAILURE in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2c916cc0a89fb7482071ab91e047db033cb865f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-23 11:34:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2dc543d2e03f824e6d58bdf033e7f6072684699c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 11:42:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5a53c7155fab4423afe8722962dca0b5 : SUCCESS in 2m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/91fa0f1d1b674126ac4fe067afca5f60 : SUCCESS in 2m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8367e40107745dab15c7fbf36df5656 : FAILURE in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/867e773df8e7486d925f796310ac4f83 : FAILURE in 2m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3f3c1f46d904831bbac7d6cdc04fb69 : FAILURE in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c4876f09d014f35aacc3612a5385c54 : FAILURE in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d67332f98ed540329cb3b102630e6744 : FAILURE in 3m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a907a13be1d2421aaac519bed0c99ce1 : SUCCESS in 5m 22s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/cc24be0e3a8c45688db2f4495714833c : FAILURE in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a869c70a42b4a88fbb29fa62cb201f6dbad6fbf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-24 00:54:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"60429dd2250365cd15dcdc69826c0a37e07d2877","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 01:21:05.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f4a64eabcfa943b49c80e49298f16ef2 : SUCCESS in 2m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58fb9726fae847abaf14980385a00ce1 : FAILURE in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ec96bedcfb4bd8bb4ae8a61582778f : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a2448b2a47547f5b62554660ccd34bb : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e299eb7515b4c56a9b9608000eb3331 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd98d0dbb8d54edb9b77070f57548bb3 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/639a2bb710844a7d80d6d4d364a42843 : SUCCESS in 7m 35s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/c29b71d04cfb404a870a0e8810adb479 : FAILURE in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f2fb1a58554a728fe2a37fd76d77d8d7d2b7fdf4","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-06-24 02:52:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6095927c7d1a170dc7bce0373e457933a33e1242","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 03:08:54.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/804342326f72419c9afd787d4a0f0ffc : SUCCESS in 2m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5118e17e12b9490ba7741640b88bc0d9 : FAILURE in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ea242e05b34a69a3b9d9e50aef55f0 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d263fffe16434707b48d0b761ed42095 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a51eb38e27784ab0b8d37b6c79926be8 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2ae7c3c9c2b41619d029faae8bcb0df : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3697665337a2421ca98b3e2d6d915cb2 : SUCCESS in 8m 29s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/c880d3dd2e7548e7bb11dbeb8a250bff : FAILURE in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a5fcb5f72a23b81e8843995fe1b93e965879a63c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-24 03:11:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8e6d47e02e3bcd6addf81b05dc8c54c3da5b654d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 03:20:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/309d169f586a406ab6da4bde4f4fc458 : SUCCESS in 2m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6d50372763b54194ac2dbf98fa6f9bc0 : SUCCESS in 2m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc31644555ce4534986108c713f35307 : FAILURE in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9edb9ad3f3b478ea6c032c1b56451f7 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/728cb46a551e459ebf3b83a52dc5acd8 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c53c90c76a414bcdae803a75caef412d : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1831073d51624a9895326009f9ff0062 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e716a48dc1745db8cbfca7d684c23f1 : SUCCESS in 7m 02s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/680d08521bf64800bb5984349f98e490 : FAILURE in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"79df2318b4c51d55f0a6830d1d561e06a0378e1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-24 03:29:14.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"41e3ff359dac532154868d2f73e89fffb52b600f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 03:41:47.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7805c0de93a9494aab8930209a1b813f : SUCCESS in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cc496b1fbda34b6da9793dab4ac45110 : FAILURE in 1m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d4b959b10d04ccc86a47f1439134b74 : FAILURE in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/246543666c8244bf90db510ea45acf52 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/083f86bf86fd4cb5bbe37c1deddb2a33 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c06c1aff2e3480e889c943dbbdb42a1 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd6298863436441984aa5ddbbe613444 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecd5588681eb4eaaa95e070eb5284885 : SUCCESS in 6m 03s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/3d8dfee752d3486fbf4d3afd025a108c : FAILURE in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"053a58fc3c89fd0318b9f0d6065ba75276204899","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-24 03:47:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d287ea5c8d87209b5a3e779bac91cdc02c5fe93b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 03:58:28.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ea76b1a1d01445faa2267a2ef0548f75 : SUCCESS in 1m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8e320b0b433049e5b429007d151edad2 : FAILURE in 2m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fe5718249e441d2b73bf5f2cb26c15f : FAILURE in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce1bab3ee5b442f8933cef102e482ee : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fa8ebce58bc4e38af9adafb3ed90731 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a8d899383164ae287684198177f2972 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ac3b8b1ce184e378c313f827cb628c7 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ed1c566e2645eda808588d9d88ea2a : SUCCESS in 5m 50s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/1e7191afa9724b97b27f4b981505ad06 : FAILURE in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e7eb0af12699392f07afeb7bde8b0d48c3cbf003","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-24 05:50:49.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7588e49d0cd7396f9af3073189143437104416ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 06:04:19.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bb3f47e3e86241bb8574aa2093a07b17 : SUCCESS in 2m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f578c78aefb14bcfb52d216f591390d8 : FAILURE in 2m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98824b3c1c324966a5915a2dcdf5cf7b : FAILURE in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c6d4afa0bd94401be79c1c38cf82785 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbc817d933a24e719f1c989013ccd93f : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d785d4151f04f999ffafef8418e76a6 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb1430970e0d4ff785de76c6f64defc6 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06a2189f44264d38ba8b29cfdde7950d : SUCCESS in 6m 33s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/9ba796f41d174393b66c0266bb2c97d4 : FAILURE in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"331cb7740cf6af7ff522baf2383847fa2a6372ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-24 06:24:04.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"24f131223eb8e204dbcae7c60edb042401be2f77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 06:36:26.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3b97ec3f010547748288a4202931e783 : SUCCESS in 2m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1b82474955994f25bd4abeaa163ca213 : FAILURE in 2m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/593736a4ce8a49578682f847e8784858 : FAILURE in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/931bd0d4f9d74b2fb75d68b8430be9a3 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b177b814b41b4c18a9230a73e4460d44 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92a50cfd85e84fc7acfb0d174bc654f3 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc2e04d8d1a44784bf2753337af423e7 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6714e0d6ebc4f04a1f47fbb7c2044c6 : SUCCESS in 6m 28s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/6b0afd08975c43469d89e4c3fbfaed5a : FAILURE in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"170299a5d05e4db52b305d155a667ab0573b4a73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-24 06:41:10.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"d448bb397acf20fe20772837f3b90b0d67217094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 07:23:09.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5ed2f7e84eb54f2896cdf897f7fb52c9 : SUCCESS in 2m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dce396d03fac40ef9b237148960c1507 : FAILURE in 2m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70cd5e3541f84ea9bc65ae234df77502 : TIMED_OUT in 40m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89a7a97e15e74f2e96d8422d1d1d46a0 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d127c12a81d40beb49656ce4de4a867 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d23eb347db24c95a1ee943741d54e29 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd705c0b6aae4726983991e4c35bf223 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf61a514fb1a45248e353beb6b7fb864 : SUCCESS in 6m 47s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/41083a2024da4ddebfca5e9f8ce6403f : FAILURE in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2afcf0aca98ed181869ee0b67f5032c8acf19d1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-24 09:56:49.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ca6c151295741ab0733cf42fd634f726899f72a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 10:44:21.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/23edf94a5a744793bb7829c9f4135f9e : SUCCESS in 2m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/301df858ab1647e887bfb08b9785e787 : SUCCESS in 2m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e16d8357a724b7c96ba26dc5a131b11 : TIMED_OUT in 41m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db12dc1b44f64a2a912766f4f5da3615 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1e6855386b141b0b055efedca846e4c : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/706930f85cfd48d986775b16be87b05d : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e85ae904f64844fd9ab3e7817cf3fe8d : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a43951187a14c429e794b264eb0aee5 : SUCCESS in 6m 18s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/23aa5b6830204f96acc172cbef320690 : FAILURE in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b9323b52c2977ce6d4af4723152507de38c183bb","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-06-27 08:24:42.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6e082052308baf5126808aa945c72d96d01816da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 09:07:47.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/da75d2e06525440c8c3a945d11f1d1b7 : SUCCESS in 2m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/00c985cb0ac6404d843e1653c162f0db : SUCCESS in 3m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32e95ed1005548c385aa176146d450b6 : TIMED_OUT in 41m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e3b8753dcf45df9d20619e51e8f5fb : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9d2e776df43422399ec2cfe2619f53d : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66dd88c92cb542afb8a61bb217e1500a : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75a1f75b6fa54134950278b5f0c4a552 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe129c29f0e74b6496e67af0501139d2 : SUCCESS in 6m 54s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/9e2e77135f5d4d8eb7543424ceb5785b : FAILURE in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"d36d2b30dd3cf782a27cc659ce89e08c8dc5bbb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-28 01:21:20.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"96463e796f706566faa12a77da6f2f73b7bef47b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 02:05:18.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/efecfa43c8b24b87975209e312f0d523 : SUCCESS in 1m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e39242eac0d54528be416054392a92df : SUCCESS in 2m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7685ff5ee97d4c119400bfcc9a502d9b : TIMED_OUT in 40m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c06a21795104f2fb3fa416dbb141333 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83ae1c8569ef41e2aa6bf3a5252b48a3 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2393a9dc94c6423b853a4d65fe2d3388 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec5af730c8cc4f9180a4a5ea4d92360d : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f91fff136614e04b9e0f038e91b7804 : SUCCESS in 5m 23s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/13417996c6854bee970d501b775bd829 : FAILURE in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"fd010c9639d8019441c13c27d16df5d7c5da10dd","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-06-29 08:21:15.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"4fbb1c77fcc46539896cce26939b5d54a1e0e93f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-06-29 08:21:35.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"1dee38e171298a1060c79361950e9cae601db9ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-29 08:50:40.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"0c49f387e77534a1a686d9d8f2fc5839e76f1a4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-29 08:54:29.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"07144e05293c283528ad9b34e5310e5ff3e70236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 09:06:39.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/becb1f2ad8814566ba2e8b183b80bc39 : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab18c5c962d5410faf89b599eab6f507 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cc2075aabbd48ca98765107c1a3dd22 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1994a4dfd6e4b62bcb05e94452c6ab5 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/045acb581b35449eb05b376d90288e13 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8dfaff373024347a67160841559a980 : SUCCESS in 6m 43s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/f0f406977f3f44ee8545911d52fbcccb : FAILURE in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"50ccac4fbee2f7780b6239f06581f5965f7c425c","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2022-06-30 00:51:53.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"9ffcf67568a626b02df146971cc8565a9a4e2140","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-06-30 02:57:12.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"5949777e03868c9d1b557557f6c9db8fe7168345","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-06-30 07:12:51.000000000","message":"Topic drop-l-c removed","accounts_in_message":[],"_revision_number":16},{"id":"7a84d12ab57a5c42271d48acfb674b1c9187cff0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 10:26:55.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"abf561c7ddc6ca0b4a4af72721668238b8e335ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 11:16:45.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/53249e2d030241e6bd057206c0a870c0 : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85067fc5f76741658afc0afb019cc72e : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e5c9743342f46a4aed3d89fcd953742 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38095b9c405a4993bd908f89ee401fa2 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4739ba371c1640e0af8a7850b2a05415 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":16},{"id":"89076ed23e5716aebc13226f3476c56cac6435a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 11:16:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"94d611d03fd3413e41bcfcff6ef0914fe1a2cfb9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 11:21:50.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb203d58a90f48d49673e4d5dad9b0bf : SUCCESS in 49s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"6b0409043f3e77d47569c3239aff0c9110940a81","revisions":{"24b476cd9b0fe3347e20012aa493ea804a37b98d":{"kind":"REWORK","_number":1,"created":"2021-12-21 10:24:54.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/1"}}},"commit":{"parents":[{"commit":"f1f7da0333bcc32a79872091099c070bd4bd0fc0","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/f1f7da0333bcc32a79872091099c070bd4bd0fc0"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/24b476cd9b0fe3347e20012aa493ea804a37b98d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/24b476cd9b0fe3347e20012aa493ea804a37b98d"}]},"branch":"refs/heads/master"},"381f3b1d3312c8d8c6fb7df2147128b81c49d715":{"kind":"REWORK","_number":2,"created":"2021-12-24 01:41:52.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/2"}}},"commit":{"parents":[{"commit":"f1f7da0333bcc32a79872091099c070bd4bd0fc0","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/f1f7da0333bcc32a79872091099c070bd4bd0fc0"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-24 01:41:19.000000000","tz":480},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/381f3b1d3312c8d8c6fb7df2147128b81c49d715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/381f3b1d3312c8d8c6fb7df2147128b81c49d715"}]},"branch":"refs/heads/master"},"a9a87fc5a700a19df72439d34b33b1da6d7b59d6":{"kind":"REWORK","_number":3,"created":"2021-12-27 07:44:49.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/3"}}},"commit":{"parents":[{"commit":"f1f7da0333bcc32a79872091099c070bd4bd0fc0","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/f1f7da0333bcc32a79872091099c070bd4bd0fc0"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-27 07:44:22.000000000","tz":480},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a9a87fc5a700a19df72439d34b33b1da6d7b59d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a9a87fc5a700a19df72439d34b33b1da6d7b59d6"}]},"branch":"refs/heads/master"},"7d7010712f0f7e219c90ff3f0ec0e2e8788dc353":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-06-17 07:11:38.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/66/822466/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/4"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2022-06-17 07:11:38.000000000","tz":0},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/7d7010712f0f7e219c90ff3f0ec0e2e8788dc353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/7d7010712f0f7e219c90ff3f0ec0e2e8788dc353"}]},"branch":"refs/heads/master","description":"Rebase"},"b1eb56afbfc9fa3ac2930f5193b3187a5b817bae":{"kind":"REWORK","_number":5,"created":"2022-06-23 11:34:39.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/5"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-23 11:34:16.000000000","tz":480},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/b1eb56afbfc9fa3ac2930f5193b3187a5b817bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/b1eb56afbfc9fa3ac2930f5193b3187a5b817bae"}]},"branch":"refs/heads/master"},"8d7a12b76edbb1466f02845dd96ab2943860d68e":{"kind":"REWORK","_number":6,"created":"2022-06-24 00:54:48.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/6"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-24 00:54:42.000000000","tz":480},"subject":"Switch task to stack-python3-zed-jobs","message":"Switch task to stack-python3-zed-jobs\n\nIn zed cycle, OpenStack projects has dropped the py3.6 and py3.7\nsupport, so we will remove them from this project.\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/8d7a12b76edbb1466f02845dd96ab2943860d68e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/8d7a12b76edbb1466f02845dd96ab2943860d68e"}]},"branch":"refs/heads/master"},"597e2e3d02e5979fcc306070d4e955341e07300f":{"kind":"REWORK","_number":7,"created":"2022-06-24 03:11:31.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/7"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-24 03:10:51.000000000","tz":480},"subject":"Switch task to stack-python3-zed-jobs","message":"Switch task to stack-python3-zed-jobs\n\nIn zed cycle, OpenStack projects has dropped the py3.6 and py3.7\nsupport, so we will remove them from this project.\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nThe conflict is caused by:\n    flake8 3.7.9 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.8 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.7 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.6 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.5 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.4 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.3 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.2 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.1 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.0 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.6.0 depends on pycodestyle\u003c2.5.0 and \u003e\u003d2.4.0\n    The user requested (constraint) pycodestyle\u003d\u003d2.0.0\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/597e2e3d02e5979fcc306070d4e955341e07300f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/597e2e3d02e5979fcc306070d4e955341e07300f"}]},"branch":"refs/heads/master"},"4ba3c9bee3f52f95f639cdd04d45add6323ebfdc":{"kind":"REWORK","_number":8,"created":"2022-06-24 03:29:14.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/8"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-24 03:28:32.000000000","tz":480},"subject":"Switch task to stack-python3-zed-jobs","message":"Switch task to stack-python3-zed-jobs\n\nIn zed cycle, OpenStack projects has dropped the py3.6 and py3.7\nsupport, so we will remove them from this project.\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nThe conflict is caused by:\n    flake8 3.7.9 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.8 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.7 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.6 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.5 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.4 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.3 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.2 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.1 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.0 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.6.0 depends on pycodestyle\u003c2.5.0 and \u003e\u003d2.4.0\n    The user requested (constraint) pycodestyle\u003d\u003d2.0.0\n\nThe conflict is caused by:\n    The user requested requests\u003e\u003d2.14.2\n    sahara 10.0.0.0b1 depends on requests\u003e\u003d2.14.2\n    sphinx 2.0.0 depends on requests\u003e\u003d2.5.0\n    keystoneauth1 3.11.2 depends on requests\u003e\u003d2.14.2\n    keystonemiddleware 5.3.0 depends on requests\u003e\u003d2.14.2\n    oslo-config 6.8.0 depends on requests\u003e\u003d2.18.0\n    The user requested (constraint) requests\u003d\u003d2.14.2\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/4ba3c9bee3f52f95f639cdd04d45add6323ebfdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/4ba3c9bee3f52f95f639cdd04d45add6323ebfdc"}]},"branch":"refs/heads/master"},"a188753dfbe74f171ab578a5464327464c5fb26d":{"kind":"REWORK","_number":9,"created":"2022-06-24 03:47:52.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/9"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-24 03:47:10.000000000","tz":480},"subject":"Switch task to stack-python3-zed-jobs","message":"Switch task to stack-python3-zed-jobs\n\nIn zed cycle, OpenStack projects has dropped the py3.6 and py3.7\nsupport, so we will remove them from this project.\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nThe conflict is caused by:\n    flake8 3.7.9 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.8 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.7 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.6 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.5 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.4 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.3 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.2 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.1 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.0 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.6.0 depends on pycodestyle\u003c2.5.0 and \u003e\u003d2.4.0\n    The user requested (constraint) pycodestyle\u003d\u003d2.0.0\n\nThe conflict is caused by:\n    The user requested requests\u003e\u003d2.14.2\n    sahara 10.0.0.0b1 depends on requests\u003e\u003d2.14.2\n    sphinx 2.0.0 depends on requests\u003e\u003d2.5.0\n    keystoneauth1 3.11.2 depends on requests\u003e\u003d2.14.2\n    keystonemiddleware 5.3.0 depends on requests\u003e\u003d2.14.2\n    oslo-config 6.8.0 depends on requests\u003e\u003d2.18.0\n    The user requested (constraint) requests\u003d\u003d2.14.2\n\nThe conflict is caused by:\n    requests 2.18.0 depends on urllib3\u003c1.22 and \u003e\u003d1.21.1\n    The user requested (constraint) urllib3\u003d\u003d1.24.1\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a188753dfbe74f171ab578a5464327464c5fb26d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a188753dfbe74f171ab578a5464327464c5fb26d"}]},"branch":"refs/heads/master"},"005eb7bbd56f9cbe0fff590f8e909a3134545648":{"kind":"REWORK","_number":10,"created":"2022-06-24 05:50:49.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/10"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-24 05:50:04.000000000","tz":480},"subject":"Switch task to stack-python3-zed-jobs","message":"Switch task to stack-python3-zed-jobs\n\nIn zed cycle, OpenStack projects has dropped the py3.6 and py3.7\nsupport, so we will remove them from this project.\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nThe conflict is caused by:\n    flake8 3.7.9 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.8 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.7 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.6 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.5 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.4 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.3 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.2 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.1 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.0 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.6.0 depends on pycodestyle\u003c2.5.0 and \u003e\u003d2.4.0\n    The user requested (constraint) pycodestyle\u003d\u003d2.0.0\n\nThe conflict is caused by:\n    The user requested requests\u003e\u003d2.14.2\n    sahara 10.0.0.0b1 depends on requests\u003e\u003d2.14.2\n    sphinx 2.0.0 depends on requests\u003e\u003d2.5.0\n    keystoneauth1 3.11.2 depends on requests\u003e\u003d2.14.2\n    keystonemiddleware 5.3.0 depends on requests\u003e\u003d2.14.2\n    oslo-config 6.8.0 depends on requests\u003e\u003d2.18.0\n    The user requested (constraint) requests\u003d\u003d2.14.2\n\nThe conflict is caused by:\n    requests 2.18.0 depends on urllib3\u003c1.22 and \u003e\u003d1.21.1\n    The user requested (constraint) urllib3\u003d\u003d1.24.1\n\nThe conflict is caused by:\n    requests 2.18.0 depends on urllib3\u003c1.22 and \u003e\u003d1.21.1\n    botocore 1.12.87 depends on urllib3\u003c1.25 and \u003e\u003d1.20; python_version \u003e\u003d \"3.4\"\n    dulwich 0.19.10 depends on urllib3\u003e\u003d1.23\n    The user requested (constraint) urllib3\u003d\u003d1.21.1\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/005eb7bbd56f9cbe0fff590f8e909a3134545648"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/005eb7bbd56f9cbe0fff590f8e909a3134545648"}]},"branch":"refs/heads/master"},"ddf2e9c994602a6e9c8fbf4f84d064c54dabaf37":{"kind":"REWORK","_number":11,"created":"2022-06-24 06:24:04.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/11"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-24 06:21:38.000000000","tz":480},"subject":"Switch task to stack-python3-zed-jobs","message":"Switch task to stack-python3-zed-jobs\n\nIn zed cycle, OpenStack projects has dropped the py3.6 and py3.7\nsupport, so we will remove them from this project.\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nThe conflict is caused by:\n    flake8 3.7.9 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.8 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.7 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.6 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.5 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.4 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.3 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.2 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.1 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.0 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.6.0 depends on pycodestyle\u003c2.5.0 and \u003e\u003d2.4.0\n    The user requested (constraint) pycodestyle\u003d\u003d2.0.0\n\nThe conflict is caused by:\n    The user requested requests\u003e\u003d2.14.2\n    sahara 10.0.0.0b1 depends on requests\u003e\u003d2.14.2\n    sphinx 2.0.0 depends on requests\u003e\u003d2.5.0\n    keystoneauth1 3.11.2 depends on requests\u003e\u003d2.14.2\n    keystonemiddleware 5.3.0 depends on requests\u003e\u003d2.14.2\n    oslo-config 6.8.0 depends on requests\u003e\u003d2.18.0\n    The user requested (constraint) requests\u003d\u003d2.14.2\n\nThe conflict is caused by:\n    requests 2.18.0 depends on urllib3\u003c1.22 and \u003e\u003d1.21.1\n    The user requested (constraint) urllib3\u003d\u003d1.24.1\n\nThe conflict is caused by:\n    requests 2.18.0 depends on urllib3\u003c1.22 and \u003e\u003d1.21.1\n    botocore 1.12.87 depends on urllib3\u003c1.25 and \u003e\u003d1.20; python_version \u003e\u003d \"3.4\"\n    dulwich 0.19.10 depends on urllib3\u003e\u003d1.23\n    The user requested (constraint) urllib3\u003d\u003d1.21.1\n\nThe conflict is caused by:\n    The user requested oslo.utils\u003e\u003d3.33.0\n    oslo-log 3.36.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-serialization 2.18.0 depends on oslo.utils\u003e\u003d3.20.0\n    sahara 10.0.0.0b1 depends on oslo.utils\u003e\u003d3.33.0\n    castellan 1.1.0 depends on oslo.utils\u003e\u003d3.33.0\n    keystonemiddleware 5.3.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-concurrency 3.29.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-db 6.0.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-messaging 9.4.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-middleware 3.37.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-service 1.36.0 depends on oslo.utils\u003e\u003d3.38.0\n    The user requested (constraint) oslo-utils\u003d\u003d3.33.0\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/ddf2e9c994602a6e9c8fbf4f84d064c54dabaf37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/ddf2e9c994602a6e9c8fbf4f84d064c54dabaf37"}]},"branch":"refs/heads/master"},"54e31644ffd085551edba2df6d59574f685d949e":{"kind":"REWORK","_number":12,"created":"2022-06-24 06:41:10.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/12"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-24 06:40:07.000000000","tz":480},"subject":"Switch task to stack-python3-zed-jobs","message":"Switch task to stack-python3-zed-jobs\n\nIn zed cycle, OpenStack projects has dropped the py3.6 and py3.7\nsupport, so we will remove them from this project.\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nThe conflict is caused by:\n    flake8 3.7.9 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.8 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.7 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.6 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.5 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.4 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.3 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.2 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.1 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.0 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.6.0 depends on pycodestyle\u003c2.5.0 and \u003e\u003d2.4.0\n    The user requested (constraint) pycodestyle\u003d\u003d2.0.0\n\nThe conflict is caused by:\n    The user requested requests\u003e\u003d2.14.2\n    sahara 10.0.0.0b1 depends on requests\u003e\u003d2.14.2\n    sphinx 2.0.0 depends on requests\u003e\u003d2.5.0\n    keystoneauth1 3.11.2 depends on requests\u003e\u003d2.14.2\n    keystonemiddleware 5.3.0 depends on requests\u003e\u003d2.14.2\n    oslo-config 6.8.0 depends on requests\u003e\u003d2.18.0\n    The user requested (constraint) requests\u003d\u003d2.14.2\n\nThe conflict is caused by:\n    requests 2.18.0 depends on urllib3\u003c1.22 and \u003e\u003d1.21.1\n    The user requested (constraint) urllib3\u003d\u003d1.24.1\n\nThe conflict is caused by:\n    requests 2.18.0 depends on urllib3\u003c1.22 and \u003e\u003d1.21.1\n    botocore 1.12.87 depends on urllib3\u003c1.25 and \u003e\u003d1.20; python_version \u003e\u003d \"3.4\"\n    dulwich 0.19.10 depends on urllib3\u003e\u003d1.23\n    The user requested (constraint) urllib3\u003d\u003d1.21.1\n\nThe conflict is caused by:\n    The user requested oslo.utils\u003e\u003d3.33.0\n    oslo-log 3.36.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-serialization 2.18.0 depends on oslo.utils\u003e\u003d3.20.0\n    sahara 10.0.0.0b1 depends on oslo.utils\u003e\u003d3.33.0\n    castellan 1.1.0 depends on oslo.utils\u003e\u003d3.33.0\n    keystonemiddleware 5.3.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-concurrency 3.29.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-db 6.0.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-messaging 9.4.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-middleware 3.37.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-service 1.36.0 depends on oslo.utils\u003e\u003d3.38.0\n    The user requested (constraint) oslo-utils\u003d\u003d3.33.0\n\nThe conflict is caused by:\n    oslo-messaging 9.4.0 depends on amqp\u003e\u003d2.4.0\n    kombu 5.0.2 depends on amqp\u003c6.0.0 and \u003e\u003d5.0.0\n    The user requested (constraint) amqp\u003d\u003d2.4.1\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/54e31644ffd085551edba2df6d59574f685d949e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/54e31644ffd085551edba2df6d59574f685d949e"}]},"branch":"refs/heads/master"},"944a47e9da6454a867ea3dc39388536fced3e8cd":{"kind":"REWORK","_number":13,"created":"2022-06-24 09:56:49.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/13"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-24 09:56:18.000000000","tz":480},"subject":"Switch task to stack-python3-zed-jobs","message":"Switch task to stack-python3-zed-jobs\n\nIn zed cycle, OpenStack projects has dropped the py3.6 and py3.7\nsupport, so we will remove them from this project.\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nThe conflict is caused by:\n    flake8 3.7.9 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.8 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.7 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.6 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.5 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.4 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.3 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.2 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.1 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.0 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.6.0 depends on pycodestyle\u003c2.5.0 and \u003e\u003d2.4.0\n    The user requested (constraint) pycodestyle\u003d\u003d2.0.0\n\nThe conflict is caused by:\n    The user requested requests\u003e\u003d2.14.2\n    sahara 10.0.0.0b1 depends on requests\u003e\u003d2.14.2\n    sphinx 2.0.0 depends on requests\u003e\u003d2.5.0\n    keystoneauth1 3.11.2 depends on requests\u003e\u003d2.14.2\n    keystonemiddleware 5.3.0 depends on requests\u003e\u003d2.14.2\n    oslo-config 6.8.0 depends on requests\u003e\u003d2.18.0\n    The user requested (constraint) requests\u003d\u003d2.14.2\n\nThe conflict is caused by:\n    requests 2.18.0 depends on urllib3\u003c1.22 and \u003e\u003d1.21.1\n    The user requested (constraint) urllib3\u003d\u003d1.24.1\n\nThe conflict is caused by:\n    requests 2.18.0 depends on urllib3\u003c1.22 and \u003e\u003d1.21.1\n    botocore 1.12.87 depends on urllib3\u003c1.25 and \u003e\u003d1.20; python_version \u003e\u003d \"3.4\"\n    dulwich 0.19.10 depends on urllib3\u003e\u003d1.23\n    The user requested (constraint) urllib3\u003d\u003d1.21.1\n\nThe conflict is caused by:\n    The user requested oslo.utils\u003e\u003d3.33.0\n    oslo-log 3.36.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-serialization 2.18.0 depends on oslo.utils\u003e\u003d3.20.0\n    sahara 10.0.0.0b1 depends on oslo.utils\u003e\u003d3.33.0\n    castellan 1.1.0 depends on oslo.utils\u003e\u003d3.33.0\n    keystonemiddleware 5.3.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-concurrency 3.29.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-db 6.0.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-messaging 9.4.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-middleware 3.37.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-service 1.36.0 depends on oslo.utils\u003e\u003d3.38.0\n    The user requested (constraint) oslo-utils\u003d\u003d3.33.0\n\nThe conflict is caused by:\n    oslo-messaging 9.4.0 depends on amqp\u003e\u003d2.4.0\n    kombu 5.0.2 depends on amqp\u003c6.0.0 and \u003e\u003d5.0.0\n    The user requested (constraint) amqp\u003d\u003d2.4.1\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/944a47e9da6454a867ea3dc39388536fced3e8cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/944a47e9da6454a867ea3dc39388536fced3e8cd"}]},"branch":"refs/heads/master"},"a768f63917b12e43e20b43856d5225de836de8db":{"kind":"REWORK","_number":14,"created":"2022-06-28 01:21:20.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/14"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-28 01:19:34.000000000","tz":480},"subject":"Switch task to stack-python3-zed-jobs","message":"Switch task to stack-python3-zed-jobs\n\nIn zed cycle, OpenStack projects has dropped the py3.6 and py3.7\nsupport, so we will remove them from this project.\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nThe conflict is caused by:\n    flake8 3.7.9 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.8 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.7 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.6 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.5 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.4 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.3 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.2 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.1 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.7.0 depends on pycodestyle\u003c2.6.0 and \u003e\u003d2.5.0\n    flake8 3.6.0 depends on pycodestyle\u003c2.5.0 and \u003e\u003d2.4.0\n    The user requested (constraint) pycodestyle\u003d\u003d2.0.0\n\nThe conflict is caused by:\n    The user requested requests\u003e\u003d2.14.2\n    sahara 10.0.0.0b1 depends on requests\u003e\u003d2.14.2\n    sphinx 2.0.0 depends on requests\u003e\u003d2.5.0\n    keystoneauth1 3.11.2 depends on requests\u003e\u003d2.14.2\n    keystonemiddleware 5.3.0 depends on requests\u003e\u003d2.14.2\n    oslo-config 6.8.0 depends on requests\u003e\u003d2.18.0\n    The user requested (constraint) requests\u003d\u003d2.14.2\n\nThe conflict is caused by:\n    requests 2.18.0 depends on urllib3\u003c1.22 and \u003e\u003d1.21.1\n    The user requested (constraint) urllib3\u003d\u003d1.24.1\n\nThe conflict is caused by:\n    requests 2.18.0 depends on urllib3\u003c1.22 and \u003e\u003d1.21.1\n    botocore 1.12.87 depends on urllib3\u003c1.25 and \u003e\u003d1.20; python_version \u003e\u003d \"3.4\"\n    dulwich 0.19.10 depends on urllib3\u003e\u003d1.23\n    The user requested (constraint) urllib3\u003d\u003d1.21.1\n\nThe conflict is caused by:\n    The user requested oslo.utils\u003e\u003d3.33.0\n    oslo-log 3.36.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-serialization 2.18.0 depends on oslo.utils\u003e\u003d3.20.0\n    sahara 10.0.0.0b1 depends on oslo.utils\u003e\u003d3.33.0\n    castellan 1.1.0 depends on oslo.utils\u003e\u003d3.33.0\n    keystonemiddleware 5.3.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-concurrency 3.29.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-db 6.0.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-messaging 9.4.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-middleware 3.37.0 depends on oslo.utils\u003e\u003d3.33.0\n    oslo-service 1.36.0 depends on oslo.utils\u003e\u003d3.38.0\n    The user requested (constraint) oslo-utils\u003d\u003d3.33.0\n\nThe conflict is caused by:\n    oslo-messaging 9.4.0 depends on amqp\u003e\u003d2.4.0\n    kombu 5.0.2 depends on amqp\u003c6.0.0 and \u003e\u003d5.0.0\n    The user requested (constraint) amqp\u003d\u003d2.4.1\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a768f63917b12e43e20b43856d5225de836de8db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a768f63917b12e43e20b43856d5225de836de8db"}]},"branch":"refs/heads/master"},"4de2b4d8097b033305613075810c22b6a082d394":{"kind":"REWORK","_number":15,"created":"2022-06-29 08:50:40.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/15"}}},"commit":{"parents":[{"commit":"70677e452bc8f7ed9103a84f50de085f7a8d99c6","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/70677e452bc8f7ed9103a84f50de085f7a8d99c6"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-29 08:47:53.000000000","tz":480},"subject":"Remove py36 and py37","message":"Remove py36 and py37\n\nIn zed cycle, OpenStack projects has dropped the py3.6 and py3.7\nsupport, so we will remove them from this project.\n\nAnd add py38 and py39 support in setup.cfg file.\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/4de2b4d8097b033305613075810c22b6a082d394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/4de2b4d8097b033305613075810c22b6a082d394"}]},"branch":"refs/heads/master"},"6b0409043f3e77d47569c3239aff0c9110940a81":{"kind":"REWORK","_number":16,"created":"2022-06-29 08:54:29.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/66/822466/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/66/822466/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/66/822466/16"}}},"commit":{"parents":[{"commit":"70677e452bc8f7ed9103a84f50de085f7a8d99c6","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/70677e452bc8f7ed9103a84f50de085f7a8d99c6"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-29 08:54:08.000000000","tz":480},"subject":"Remove py36 and py37","message":"Remove py36 and py37\n\nIn zed cycle, OpenStack projects has dropped the py3.6 and py3.7\nsupport, so we will remove them from this project.\n\nAnd add py38 and py39 support in setup.cfg file.\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: I7805ba2ff67335d23917de8b04d6aeaeff83120f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/6b0409043f3e77d47569c3239aff0c9110940a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/6b0409043f3e77d47569c3239aff0c9110940a81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}}]}],"submit_requirements":[]}
