)]}'
{"id":"openstack%2Fsahara-plugin-cdh~840120","triplet_id":"openstack%2Fsahara-plugin-cdh~master~I38c6140decdba5c02e605265eeffaf89def762a2","project":"openstack/sahara-plugin-cdh","branch":"master","topic":"drop-l-c","hashtags":[],"change_id":"I38c6140decdba5c02e605265eeffaf89def762a2","subject":"Drop lower-constraints.txt and its testing","status":"ABANDONED","created":"2022-05-01 18:03:57.000000000","updated":"2024-05-15 05:08:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b046787a8473c4d1d87dddf295407a37090e5b16","_number":840120,"virtual_id_number":840120,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-01 18:22:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-01 18:22:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7a78fad4a149e64e78eced8a2d551985e45555b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-01 18:03:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"277ee6fb7e4b1477a923dc421e99ea1f7e7b33f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 18:22:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bf1edee4d9644623afc1709634859e4a : SUCCESS in 1m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcbdb321b1f94f4ea012747ee407c895 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cc96ed9491c4cec879248be1310e94a : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/459e2562369d44109bfd77e4347eb9a0 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc12509dcb3c4a5cb7c340ed1d049015 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597810a2c3694713b28dd63e16df1397 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cf40ac537fd4c54b5de162188b3d782 : SUCCESS in 3m 34s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/261809c5bfad419dba6d6fee4f98af02 : FAILURE in 3m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b046787a8473c4d1d87dddf295407a37090e5b16","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-15 05:08:20.000000000","message":"Abandoned\n\nsahara project has been retired https://review.opendev.org/c/openstack/governance/+/919374","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bbc0f7e4e75ec1a421f048806c1e6f66db1730f4","revisions":{"bbc0f7e4e75ec1a421f048806c1e6f66db1730f4":{"kind":"REWORK","_number":1,"created":"2022-05-01 18:03:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/20/840120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/20/840120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/20/840120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/20/840120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/20/840120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/20/840120/1"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 17:54:47.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 18:03:56.000000000","tz":-300},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I38c6140decdba5c02e605265eeffaf89def762a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/bbc0f7e4e75ec1a421f048806c1e6f66db1730f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/bbc0f7e4e75ec1a421f048806c1e6f66db1730f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
