)]}'
{"id":"openstack%2Fsahara-plugin-cdh~846983","triplet_id":"openstack%2Fsahara-plugin-cdh~master~I9e52deda308339f7b0f1f8138dcf0345ccaa37cb","project":"openstack/sahara-plugin-cdh","branch":"master","attention_set":{},"removed_from_attention_set":{"34375":{"account":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"last_update":"2024-05-15 05:08:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9e52deda308339f7b0f1f8138dcf0345ccaa37cb","subject":"Bump oslo.* and pbr version to support \u003e\u003dpy38","status":"ABANDONED","created":"2022-06-24 01:41:41.000000000","updated":"2024-05-15 05:08:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"df9dd947a6adfaaba270ef3b16ee421b95edcd8d","_number":846983,"virtual_id_number":846983,"owner":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-24 02:33:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 01:56:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c01a5d1fd36bd6ab15ad4158610256547e040b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-24 01:41:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08a6684b9737edf1ce3b5b76820ac27429eda96d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 01:56:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2a7c9a377a724b04acdb0c51261e5620 : SUCCESS in 2m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f74fd8733fd74c09ae46758c59a37d13 : SUCCESS in 1m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01f9cfe8a9a94612ba18bf4bffb11a80 : FAILURE in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ebb85b0fac74ed486999dad485aa9bb : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecc7bce22469495981b5af0de8fe0fd3 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7e8650e6f164149b328f0527e99c5cd : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0816e8f734140fb9b8b38461b446ae3 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0dcc3139814e0098f0e4b40fd799c3 : SUCCESS in 6m 48s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/f4faa6ade80141a885d97cda8a3cd5c7 : FAILURE in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bfdbceca4029d0103a2c4da697959dfe7a5401b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-06-24 02:10:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99130f4e8fad29191265d6b936e99f54d0a9d126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 02:33:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7365cd2cd6e5404bae3ccece1390a972 : SUCCESS in 2m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f59fa5ae7ae34d91bb5788f2c84a7b5c : FAILURE in 3m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24d1c6d06dfd485897f966571ac16c60 : FAILURE in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3fe39212a8242ba80ed4a764bad8a15 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bd7f42f611442689734e20bbf8713ed : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efa2466951184de689c1566b165a4871 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71ae7ffd038446c3bb8983b822d8613e : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10a6dbc85e94558bc569152c2f89a41 : SUCCESS in 8m 13s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/fc87260e6aa54a829038a74989aa28e4 : FAILURE in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df9dd947a6adfaaba270ef3b16ee421b95edcd8d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-15 05:08:23.000000000","message":"Abandoned\n\nsahara project has been retired https://review.opendev.org/c/openstack/governance/+/919374","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ba1473c67b952404fc206bfc9a3a0a5486160f71","revisions":{"61759976ad1a41a326e0b87dea9853c7ce2651e2":{"kind":"REWORK","_number":1,"created":"2022-06-24 01:41:41.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/83/846983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/83/846983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/83/846983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/83/846983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/83/846983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/83/846983/1"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-24 01:36:44.000000000","tz":480},"subject":"Bump oslo.* and pbr version to support \u003e\u003dpy38","message":"Bump oslo.* and pbr version to support \u003e\u003dpy38\n\nIn zed cycle, OpenStack projects and oslo lib\nhas dropped the py3.6 and py3.7 support\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nI also decrepated oslo-log \u003c 5.0.0, so we should bump oslo.log\u003e\u003d5.0.0.\n\nBecause of oslo.log\u003e\u003d5.0.0 depends on oslo.i18n\u003e\u003d3.20.0 as below, so we\nalso need to upgrade oslo.i18n.\n\nChange-Id: I9e52deda308339f7b0f1f8138dcf0345ccaa37cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/61759976ad1a41a326e0b87dea9853c7ce2651e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/61759976ad1a41a326e0b87dea9853c7ce2651e2"}]},"branch":"refs/heads/master"},"ba1473c67b952404fc206bfc9a3a0a5486160f71":{"kind":"REWORK","_number":2,"created":"2022-06-24 02:10:39.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/83/846983/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/83/846983/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/83/846983/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/83/846983/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/83/846983/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/83/846983/2"}}},"commit":{"parents":[{"commit":"c1808107337911248325b815d61c5ea363076633","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c1808107337911248325b815d61c5ea363076633"}]}],"author":{"name":"hejunli","email":"18438695326@163.com","date":"2021-12-21 10:23:50.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-06-24 02:09:48.000000000","tz":480},"subject":"Bump oslo.* and pbr version to support \u003e\u003dpy38","message":"Bump oslo.* and pbr version to support \u003e\u003dpy38\n\nIn zed cycle, OpenStack projects and oslo lib\nhas dropped the py3.6 and py3.7 support\n\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nI also decrepated oslo-log \u003c 5.0.0, so we should bump oslo.log\u003e\u003d5.0.0.\n\nBecause of oslo.log\u003e\u003d5.0.0 depends on oslo.i18n\u003e\u003d3.20.0 as below, so we\nalso need to upgrade oslo.i18n.\n\nThe conflict is caused by:\n    The user requested oslo.utils\u003e\u003d3.33.0\n    oslo-log 5.0.0 depends on oslo.utils\u003e\u003d3.36.0\n    The user requested (constraint) oslo-utils\u003d\u003d3.33.0\n\nChange-Id: I9e52deda308339f7b0f1f8138dcf0345ccaa37cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/ba1473c67b952404fc206bfc9a3a0a5486160f71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/ba1473c67b952404fc206bfc9a3a0a5486160f71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
