)]}'
{"id":"openstack%2Fsahara-plugin-cdh~847742","triplet_id":"openstack%2Fsahara-plugin-cdh~master~Ie977b7a1dde588b1be4cae9862f11a76b7d916b5","project":"openstack/sahara-plugin-cdh","branch":"master","attention_set":{},"removed_from_attention_set":{"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-06-29 08:24:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie977b7a1dde588b1be4cae9862f11a76b7d916b5","subject":"wip:remove pycodestyle from lower-constraints.txt","status":"ABANDONED","created":"2022-06-29 05:47:36.000000000","updated":"2022-06-29 08:24:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c633f01f222cb321853dd57b63856371a883e581","_number":847742,"virtual_id_number":847742,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-29 08:05:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-29 06:35:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7784a537c984920920222be3d4272ba4b6e7a3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-06-29 05:47:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"342b540869a0ca8941c129f297b3a667c1cc9658","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 06:35:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/932cfffc6f8f4bf7b8e1a7780e633676 : SUCCESS in 1m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ada64065201849f481ff41c961bb28c5 : TIMED_OUT in 40m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e0ed42406454411a0615c4cda3d4bdf : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/642ba3bfc3e44dd4a47bee48725392dd : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5eef58e8f1648d9a6faa5f8b72d308d : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e992b387bcac4fb396f60a6d201aac42 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df9058a923c4f648efa42f04caf2373 : SUCCESS in 6m 52s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/16402c7def2b405488c2286e9e2635ec : FAILURE in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d13a74c26f35b1bb2a516011a1581e0bf8047236","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-06-29 07:18:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"03ccc811cefad64ebe03ccbfd8479fa03c091240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 08:05:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3f81fe71b65d469a9c4f359946be0331 : SUCCESS in 2m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3419fb3ef16749819636f22f791ff330 : TIMED_OUT in 41m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39ccd0b6ece948259290d279e49a8ca4 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3ad79b137b14eb2b05388a727c8b51c : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0238b5c29ddb4e8190fd5e515f1ee5b6 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7de8aa39b70942589cdd9b27557e4759 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7817b8ae299c41738b9338a13cd7fe60 : SUCCESS in 6m 32s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/17286a5ff8da41c6ade6f529258ce227 : FAILURE in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c633f01f222cb321853dd57b63856371a883e581","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-06-29 08:24:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de81c48cda16e2d773b60eb92df1d9e4c07afd8a","revisions":{"9d3a2d8b114647e33cab0d24cf9d025815b42ed1":{"kind":"REWORK","_number":1,"created":"2022-06-29 05:47:36.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/42/847742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/42/847742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/42/847742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/42/847742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/42/847742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/42/847742/1"}}},"commit":{"parents":[{"commit":"a768f63917b12e43e20b43856d5225de836de8db","subject":"Switch task to stack-python3-zed-jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a768f63917b12e43e20b43856d5225de836de8db"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-06-29 05:47:01.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-06-29 05:47:26.000000000","tz":480},"subject":"wip:remove pycodestyle from lower-constraints.txt","message":"wip:remove pycodestyle from lower-constraints.txt\n\nChange-Id: Ie977b7a1dde588b1be4cae9862f11a76b7d916b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/9d3a2d8b114647e33cab0d24cf9d025815b42ed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/9d3a2d8b114647e33cab0d24cf9d025815b42ed1"}]},"branch":"refs/heads/master"},"de81c48cda16e2d773b60eb92df1d9e4c07afd8a":{"kind":"REWORK","_number":2,"created":"2022-06-29 07:18:21.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/42/847742/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/42/847742/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/42/847742/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/42/847742/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/42/847742/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/42/847742/2"}}},"commit":{"parents":[{"commit":"a768f63917b12e43e20b43856d5225de836de8db","subject":"Switch task to stack-python3-zed-jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a768f63917b12e43e20b43856d5225de836de8db"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-06-29 05:47:01.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-06-29 07:18:12.000000000","tz":480},"subject":"wip:remove pycodestyle from lower-constraints.txt","message":"wip:remove pycodestyle from lower-constraints.txt\n\nChange-Id: Ie977b7a1dde588b1be4cae9862f11a76b7d916b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/de81c48cda16e2d773b60eb92df1d9e4c07afd8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/de81c48cda16e2d773b60eb92df1d9e4c07afd8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
