)]}'
{"id":"openstack%2Fsahara-plugin-cdh~857413","triplet_id":"openstack%2Fsahara-plugin-cdh~stable%2Fzed~I05cae500e03b20a6fca89dc8896f829ed9802523","project":"openstack/sahara-plugin-cdh","branch":"stable/zed","topic":"create-zed","hashtags":[],"change_id":"I05cae500e03b20a6fca89dc8896f829ed9802523","subject":"Update .gitreview for stable/zed","status":"MERGED","created":"2022-09-13 12:35:08.000000000","updated":"2022-09-13 17:55:05.000000000","submitted":"2022-09-13 17:54:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857413-create-zed","meta_rev_id":"c5aef0959ff6e02691d32d0c601929d5d233f6ae","_number":857413,"virtual_id_number":857413,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-13 17:54:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-09-13 17:45:40.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-09-13 17:45:40.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-13 13:01:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-13 17:45:40.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"a4858c260118952e3498b892b5f4d5888b1f185c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-09-13 12:35:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15fc5c1ef0569b4089ded6c340d2cb5678691aa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 13:01:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/854591872cbe460d8bd6ab47ee4879ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67ba5a29c1f74977a33a27b397bf0f8a : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85de5cf464f04b97ab645bb066d753cb : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89bbe9a1603344ae9c38c4a2a227b4d3 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a4bcec515ef477c9524a4b6550d7f83 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e9828906bb46868a9c98d8f1c6db5b : SUCCESS in 5m 52s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/dcf9b6ad979e4490adff366c3c851d4f : FAILURE in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea11830b4e50367a9736db18969b07e51fddb581","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-13 17:45:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2397f98316e7741073d68890c155cbb595bcacfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 17:46:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"380a7abe234365dcff9287bd9ae15211ed0e006e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 17:54:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fe26f1bf0bc4bc5bb696c6c6a2482b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb06a24ce774b43b9469eacd764f31b : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26549acf8ccc4a9f9940f7eb0b5faab0 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23355f1677d8461f92f69cd93e0205e5 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ce1f41e4354e3fa724d010ea9b316e : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":1},{"id":"64f84f90021834e5edac5f8ead89fae975d77f9f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 17:54:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c5aef0959ff6e02691d32d0c601929d5d233f6ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 17:55:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec87a2633f2649c7894d6e66633938b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7de90c24dc8f4d2fad3bf4a799e6f656 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd1bf5465d7898e188db39b151d273fe64ba51f1","revisions":{"bd1bf5465d7898e188db39b151d273fe64ba51f1":{"kind":"REWORK","_number":1,"created":"2022-09-13 12:35:08.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/13/857413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/13/857413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/13/857413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/13/857413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/13/857413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/13/857413/1"}}},"commit":{"parents":[{"commit":"8f8e9fde6c371dd68954695c67bf91121d71076c","subject":"Remove Python 2 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/8f8e9fde6c371dd68954695c67bf91121d71076c"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-13 12:35:07.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-13 12:35:07.000000000","tz":0},"subject":"Update .gitreview for stable/zed","message":"Update .gitreview for stable/zed\n\nChange-Id: I05cae500e03b20a6fca89dc8896f829ed9802523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/bd1bf5465d7898e188db39b151d273fe64ba51f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/bd1bf5465d7898e188db39b151d273fe64ba51f1"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[]}
