)]}'
{"id":"openstack%2Fsahara-plugin-cdh~894005","triplet_id":"openstack%2Fsahara-plugin-cdh~master~I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971","project":"openstack/sahara-plugin-cdh","branch":"master","attention_set":{},"removed_from_attention_set":{"29423":{"account":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"last_update":"2023-10-17 03:18:34.000000000","reason":"\u003cGERRIT_ACCOUNT_29423\u003e replied on the change","reason_account":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}}},"hashtags":[],"change_id":"I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971","subject":"Fixes the several problems/errors caused by tox","status":"MERGED","created":"2023-09-07 06:36:39.000000000","updated":"2023-10-17 03:31:53.000000000","submitted":"2023-10-17 03:30:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894005","meta_rev_id":"3eed116211c087c2de3120fc65e48fb6d35255d3","_number":894005,"virtual_id_number":894005,"owner":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 03:30:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-17 03:18:34.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-17 03:18:34.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 06:45:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a31b0b5d274f69d92a37432cd02268cb4b60a20d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-07 06:36:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"900e8872ad436d824c5575062906b7275b1926a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 06:45:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7438765b9db4c2fb25e45dac22e0f61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec6d83765bc47e69d8836920a4558b3 : FAILURE in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f90c717415064419836d2c168d8251de : FAILURE in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7db0fae873f6458fb0f007e10aba6c82 : FAILURE in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87719b7fe75543a19dc6be329f030874 : FAILURE in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c92dff2a82d34de69191f9176067e8ef : FAILURE in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c943887de98b4be58b263c2398c01b15 : SUCCESS in 2m 58s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/7ad5f05c107144a4b564d92c6ad72b74 : FAILURE in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b85f0e323174009331f3658d20b449048ac460b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-07 06:51:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"15842c0077e167efd8d086807556c81a92ff287d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 07:01:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e7f3ed6f1494642ba21055b9450a993\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22f61ace86bb4a9698be522b2bebbd6b : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe9c2e6cfc1a4b989bd4eef5635ad711 : FAILURE in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9687d648282b48b480a189da0c79b421 : FAILURE in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2ad71ea255244ddbfd193d702ca06d6 : FAILURE in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/869e3a69619b48bc9a511d0d983cafe6 : FAILURE in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75e67a0973054810ac2f49d201e2d455 : SUCCESS in 3m 56s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/5c086bff0f6d4751b0d6f10e3fce7560 : FAILURE in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0080b529cee447aac22cb442cb5a55e9cbb2cf5","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-07 08:14:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"411b7b5321dff6ca933cd30d8b8a1a666a71551e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 08:25:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b619c67eef93447eb09d8164f5160db3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c1e2309647d444c9884915382636ad5 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9f56e70e2d047039c92cc107d545485 : FAILURE in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be491deed0a946d4b1dcd548e8c82858 : FAILURE in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a88e2d5e08964335bd3694304645a5d5 : FAILURE in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5004580d47764ee388a9ed4b74702d69 : FAILURE in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/618d39fbb1ea4c8fa2ae6b4761194749 : SUCCESS in 3m 55s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/e72040ca44eb41d1822978ffed622490 : FAILURE in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"330b8cd99de9026dc66cc97d9c8023a99bd26e93","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-08 00:21:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c719683ca1139817b4a32b8a167e69abbcd5805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 00:31:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/575b596d39b24dd996f9311a18320ada\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9570f31ff16443f793ca59012cad77a4 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20c10261beb94fb89a44ada93a1f66db : FAILURE in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa5c76dfb38d44ff8b7b8bbeb4ff6d27 : FAILURE in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5827e1431d374bbb9836bc1b01219059 : FAILURE in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff052cc29c2d4fc7b8f44dc298c88f97 : FAILURE in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a3a17fa43e24fa78ecbd850d0dc0482 : SUCCESS in 2m 57s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/6aba5a82eaab4ce38e7635bfa4a0351b : FAILURE in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05265acf5a8c13012418c94b693859c028817493","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-08 09:06:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"044d4887b8d22e44fa40a2d689c28ee61992fc0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:18:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59fbbabb30d64d47a46602061d5f1e50\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc3f6c1f697d48db8598a97b70c1c0a1 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a43b77f069dc4835826022cde3999609 : FAILURE in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9872da25324545249facae0cbedc4d97 : FAILURE in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc21ff35af3a4951a2a26a404949a3cd : FAILURE in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20469ba3fca94e219ddd7580d905f3c7 : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54af866f7b6149619fe17b3bdf033e3e : SUCCESS in 3m 11s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/737e1326cda3439f8e7a0c3164346cc7 : FAILURE in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ca81e050b7388e434e9283982bc26f8643d083d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-08 10:43:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c377b95654f9fb96b493d393df926ab387f14f68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 10:53:42.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c62a1fe7f0734e7eb3938d76e323fe41\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4eb1fa5ec9540819b3094b7c152d130 : FAILURE in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3166a1ef31a440b9b94a6398d7433d54 : FAILURE in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7aaa1d8662c4588ae81f0216466555d : FAILURE in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3c24f0093d54c139b221732fa8ec2dc : FAILURE in 3m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98326de5882a42a4867abf2f930e73a4 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c36d0f63d7d848a89a2219c07aedbb61 : SUCCESS in 3m 20s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/c2406e964486431e960df24136cf4b13 : FAILURE in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"150e569aa32bfe1ff406d70e05364e05c703d326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-08 10:59:22.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"911d6a260ff1506f482b50452f3bd5481f03b4b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:06:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce40883d470e4cc0bdfb3e00b171073d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bae23a3488414c5a8efaee16a4c4efad : SUCCESS in 1m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/280f07d3b4df41a8a489b969cb21732f : FAILURE in 1m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e0894b868bd4befb468cdb6c77d69a4 : FAILURE in 1m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e21f214fbcc647188a37afcb139da0fa : FAILURE in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31e04192336c474c8f9e4030087f8b15 : FAILURE in 2m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51047016eea74c2c9667eacd55e79364 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d72d87c7fdb34b95b05955b9dea92c78 : SUCCESS in 2m 54s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/893db09afdf345a9936e176bbde43782 : FAILURE in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3d7db7d18337191b66a0356e1d89bb3a0c8b4c35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-11 00:15:24.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3dbecdd3eba875ffc65e9c5f4914a641007635ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 00:26:53.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a77be526ed864557ac8233bc4d65a360\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/93fa020780c94c9690576b9a010a415e : SUCCESS in 1m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26c898a51547433ca844bc9075df21bc : FAILURE in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da047ce934284eedadb410b44050fe07 : FAILURE in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/309eab3472744ea4a57f9370cde1f441 : FAILURE in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e20203d614604dcc93490134a9d9752b : FAILURE in 3m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9eefc1992ea4a57ae166536355c5b40 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a050409915ff4138878c1086d53e1db3 : SUCCESS in 2m 43s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/ea3f7c41070242b795d094b3d9a82c54 : FAILURE in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"624960fbf490fb1e51c68c484db5036cd3fe0b16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-11 00:30:55.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"a87345b393a921b60d5f05b5ca7d207da2a00cce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 00:42:08.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57637bf110c64a218326f011d44fd46f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e5acf76ef6fc4b288e0a6b191d95a1d9 : SUCCESS in 1m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6b56bf0163b4814ad8995262ecafb7b : FAILURE in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d1846375d2a4b60932df3f308814168 : FAILURE in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f43f74c37f14e6aa15182aafe08c305 : FAILURE in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eac2bf69fa8d4c7499e775ba5cc056c9 : FAILURE in 3m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bda50f13de24b3e97713a1b026a9dfa : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/615f84a98bb64f9eaabc480281b3e086 : SUCCESS in 3m 20s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/ccf2465a7bc44e7793c1cd076a43d666 : FAILURE in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"197d4dc574b8f1d4ac9ba5742c5f5c68518e5bef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-11 01:09:18.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"ba2b6e5a1eb0f2e24233933a12e2fb12ec5727be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 01:22:37.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4d7ee768f3c415e8732adab26ee7d1f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4848432282d247fb94e694d5206436fb : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2635cd861274a6e9e681cbe59105e23 : FAILURE in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e093d5426dc24be79498f1176d226ae7 : FAILURE in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/027d2bb418f6408c8fa83d79627e86a0 : FAILURE in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3940d33c76f4e47aa5ff67a1a5c70b9 : FAILURE in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5a1153bd4264543987b541a2c9654cc : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6220aa11107d4b5aba0db7472fcd57df : SUCCESS in 3m 24s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/c09c48560b244cf5a212328098dbe83c : FAILURE in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c12072195df20147f2c21b76070fbfc5dc725013","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-11 02:56:00.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent tree, and commit message as Patch Set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"4fa1bdb2725111b26f0baaa16dac43316dba11a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 03:07:47.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ee5865cbdeb461b80b31d7155e3d555\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a91501e84924a6ba20d4dee46f57b56 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f09aabe490e419e832d315ef658e1e6 : FAILURE in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4a4e6fbd67541899bb6007caf9d5b8e : FAILURE in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/beb3ec6bc69b4739a12327d4ae659558 : FAILURE in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90d317b6e7b94294b30d235bd050556c : FAILURE in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153d9ca631e74c69b3861654146aacbc : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c07d58e871b4928a31fe5ad13ce35f1 : SUCCESS in 4m 06s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/a8fc1300e8f3427ca45319fd0dd16797 : FAILURE in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4ba40bf74cdc666a2ca40305ab767243a3638f77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-11 05:34:43.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"9c390e0e5f0c0baabd1b775405e20ed7e06f594b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 05:45:35.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cbe4dc7a937472c8f0383c6b56469a7\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4a07c577436647d692065ba1601a6628 : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58342bb15895416287d1a70665551759 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f73cf85b014149159e938c5737cb0276 : FAILURE in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b0660968a6a4e488ee3136754f15d55 : FAILURE in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/493ac5053e8141ab8cb41fee6d4b5fa8 : FAILURE in 3m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da5070fd120454aa7da2a34ff7922e8 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8774fa2cf3c04369bea493d8ca74c21b : SUCCESS in 3m 24s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/e081aa9a089d4e55b926a59b787fbca0 : FAILURE in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0e2065ad13cf060cb43c44c29f76efe8da2f2dac","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-09-20 08:52:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"89467dba6da0f775921b80f8fbfbcfb936ebeb1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 09:07:55.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb7d223aa5e046e98fb76a35b89de195\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a799dea5b6924a79a23d3cc4c2b0574c : SUCCESS in 1m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5261c43d337248638574b7fb4e160735 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3277ce12a519444f8f471536058433a4 : FAILURE in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4758dc9416ae4c87873c01835d604c97 : FAILURE in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de94a4374bc04acd9114b1bb0e14fdba : FAILURE in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3080a76c7f455482134d2dec4557cd : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/113cc992431d4e26a5e375ca863b73ab : SUCCESS in 2m 52s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/aadaad05e20c404cb0c1684a286aea88 : FAILURE in 7m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"311d84c004ce678df786a74d202c2b9bcadeafd3","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-10-11 02:17:31.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4413fd1d1f4838f1573044f72fa14f1b94a8e3be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 03:13:45.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1adcdcd7e1124bbaa18cb90599b165f6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/01229a4baff040179313ef2f805a8ac7 : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7f08761a4fe4d63a112e5340e349ee9 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f32de892839c4c6db6e2646d6bb15dae : FAILURE in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14217e7735cc4215977784bfeaa483a2 : FAILURE in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2f7143fdb9b4e70bb5a230a6be70d5f : FAILURE in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9296e9696b8a4ff1a101507dcb58973d : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15bd77a37ca54a3090b99c00b17f263f : SUCCESS in 3m 30s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/9871551873a44062a8a4042ddbd915ba : FAILURE in 8m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7a751c5a2322e9ff153dbe39187d8e396b3eef14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-10-11 06:15:00.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"1b72c2b78a40f2744ae0bd06744884e2149a0dc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 06:26:36.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9205846fe8304b1a93c5342d098cf026\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b72b014a679c4112b9e101c253747dab : SUCCESS in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc986493514f4e88a461b9d36360920e : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0caa80c87fe49c4b821ac446e11a8ec : FAILURE in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75ece04a6de5441db45d03d7dd5247bc : FAILURE in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/413401dcd62d466b800e6083e53da119 : FAILURE in 3m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d40184d552e43a59a53596c83568879 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55dc835f93ef4a498d24c769de4c3c11 : SUCCESS in 3m 08s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/14f39a47e44f4f368e4b926516a3396e : FAILURE in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0ec369e79dc327aea15bb27baf263979f0ded5e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-10-11 07:23:54.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"c98d5bffc1398ce8b9b437ecda6c9f0e6e7c7388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 07:58:33.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1b194abbf6f4151a45a099fb4a7594a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/818abf4250764b8ea806cf6503d48d01 : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9884bbbdb1154e7d9d15afa0c3c224d1 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a20c37c1a4744844bc6bb23e2d3b40a5 : FAILURE in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25898d37de594ab89db3f151d6c3c8da : FAILURE in 3m 09s (non-voting)\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bc9a56c0b7a4cd0a6f5887a2faacd82 : FAILURE in 3m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/912b78a154ba47f281efc4496369ec9c : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/214148d2d9ad4b49a52c8265901373ca : SUCCESS in 3m 32s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/b2fe786ab49e4246acbf749d4325bb07 : FAILURE in 3m 52s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d0212ddccd3618a6a708cefccde41d0f7775e6c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-10-11 08:23:13.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"4a16e6e9b22096c86f597b93d6628b3ba1e1acc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 08:34:38.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9464ae74354746ce994fc000ec797b44\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/411172c9b85349f382423d1ee3b98a74 : SUCCESS in 2m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/091623f9c2f74809a2c2e184ff1ab141 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88fe644ed9ba4e3195a32574534bb411 : FAILURE in 3m 25s (non-voting)\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f0a6cc614ad48599162416004cc98f0 : FAILURE in 3m 42s (non-voting)\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af03332095384e9ba83f296558d9808f : FAILURE in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb940d7bfb21432d9d6b92bda06bdf8d : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e274a4014b6d436182e55e27045241f2 : SUCCESS in 3m 39s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/c043b38c9e5d4eac8db81977f9088f14 : FAILURE in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"337a76797783d09a9bd10a17ec11db900cc9cfeb","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-10-17 00:57:52.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1cea9bf79d96f6782cffa1c6165a341857d7f45c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 01:09:50.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/671eb12f3d5c4b25ab4e5c8fa2b2ecdf\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ea66a5ae816f4ac4b39cc2b58af4bcf2 : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c78caeb56aed44fc8715340050a76ed8 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d70859aea4340659b17ea4a3d206674 : FAILURE in 3m 42s (non-voting)\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5312d251670413abfbb7625a911b330 : FAILURE in 3m 38s (non-voting)\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f20e213378c431a89cba55302aedc7d : FAILURE in 3m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c16da09d66d0463b8e054f8e9a644744 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5739f222a224b6f9acee136cbe31419 : SUCCESS in 3m 09s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/922a1d2681674a749e095dda266c8b45 : FAILURE in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cb58610a37cb004ee411cbe3081e9a07358659e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-10-17 01:11:22.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"52f7161ed610846e12260c842a3ead437c52b2f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 01:23:23.000000000","message":"Patch Set 14: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc81e0089bfa40a0b09f849b7906b29d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/47c0fc799227494d9f7ddadfaf5ea0f1 : SUCCESS in 2m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70ec065f9c3c4241987956ba457dc95e : FAILURE in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/584909e06a3945b48578df67a4f1f08c : FAILURE in 3m 01s (non-voting)\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdbd286e18b041d59fa909475c0e2e0c : FAILURE in 3m 12s (non-voting)\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b87fbc024c547d9a0d3a8d075d2f993 : FAILURE in 3m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac3c51ab6e0d4ce78dfeaa052e4ef16d : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cda8aa86cb1a453187beaba4c9c3695e : SUCCESS in 3m 20s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/e3dee4752eef475b976f15b9c02ddcb9 : FAILURE in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0f3ff65073d95e35cfd0891d0baaf194e2100755","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-10-17 01:24:51.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"eed2c946e043affb93092325d936f6fe96521e0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 01:35:44.000000000","message":"Patch Set 14:\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb3f161d46c04377b0fb464aa4cdb5be\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f936f3baf43b48bba1592200f3b6db49 : SUCCESS in 1m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/750c883b1291427988e9fad4eb98c40f : FAILURE in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e54ab3146bea47b49bd3dd71990a2d92 : FAILURE in 5m 06s (non-voting)\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6834d184ac5b43f38782e58891b9e5a7 : FAILURE in 3m 37s (non-voting)\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e25725e1db744f20bfd5cc1fb9bdbe3e : FAILURE in 3m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb7a71f903684f2b837fee83fd5a790b : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1eba330aa9e24bf5a9c571109688488a : SUCCESS in 2m 55s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/ea48a5c6040c46949d1910c89af4840b : FAILURE in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"65231a044f3a8a5bf3cdd6fe706c41e586eab1ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-10-17 01:40:26.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"a8bf4df2fa85b907a2fde4cdfa1894a995be4917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 01:51:08.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/886c97bba82343c18e26e5de886a68a8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0577e5ccbf3c46559f49ed171b9f13cf : SUCCESS in 1m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b799de9b38647c3858b052445d4eec2 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7c2548b5d6347ceb8cb0f1832611974 : FAILURE in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbf4bdfab8144ad3b07b329a70161d4f : FAILURE in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/175de7b926484ff89dbde3deca52e9fc : FAILURE in 3m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46468c2d45c4222abb2b4ea12759ae5 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e370bd8ac6194883acafdc2e711d7075 : SUCCESS in 3m 24s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/199053ae358449f28e5a372f5d83e7f3 : FAILURE in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"550b489cb3c36fd3f9878775c2bde7dd05bb1e7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-10-17 02:04:18.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"e9c7cf071ab05af1bbf7bcc1ac234804436183dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 02:40:53.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a58366a490b40ef90baa26e8917e8e3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1aab6ff480bc4ceca095e63514c689e7 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df522c934b2e40aaa4ae6ae8bb547063 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f6539297dd74f18876ac2a5372811a8 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/269b5d411a2c4647a423a490416e6ef1 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f15298694df4d2da45a0cb4ab41ed18 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4ad392860f245a1b6df35c6b459087f : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6bfdc455979421e8bcac985aeb462d9 : SUCCESS in 3m 47s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/09d3c139ed4343a0aca5b55638ee9fce : FAILURE in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"b5af9e632514ff04352962f552998eb302f18b7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-10-17 02:49:37.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"ee051596a8493df86c1786247ea3d0fe266af554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 03:11:10.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91be1d7498ab4354a225bf3c78301b35\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bf3f7f877d6640b8a08e05017d020595 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840ee1ff3eeb4f15aff9b08d8beb99a3 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/925eccfd43fd4361a9e34bda20cd8a2d : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27e030bd31094c0fa834c21915ef3aab : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/884d8256e5dc42ae9dc09d3cdfeb4908 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf81fd18e314599b09bc86a88950e2e : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52da569cc4e544ae9a170c6f7b8d42c7 : SUCCESS in 4m 23s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/22a192138f41476cad2aac52386a1fab : FAILURE in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"a20691248b275f942853e9bc3568e5601f8fccc9","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-10-17 03:18:34.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"b5e77145f6f83ba52bc5c7ff23083d15a0ed2912","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 03:18:54.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"e6c83d7a575cfb2c39274842f8a8fc448d7ec3ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 03:30:37.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45fade06ed5244ef85bf4ebc48d29bda\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c3ca3776205c4f20b17615ac00ffe474 : SUCCESS in 1m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25eafbe07d0f4c9884a4f53fee119f87 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b0c012bca9e4b148a2eb7bec18e90ed : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87289656081147edba17157feed5e615 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e478300938d4cc7bda9d9dbb07c35fd : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee9b05bd48244deeb70217eb4680d729 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":17},{"id":"ddd12daa9b74b8b2a347f294a59c9114bb10e218","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 03:30:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"3eed116211c087c2de3120fc65e48fb6d35255d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 03:31:53.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa199d2352fb4856aadf5630f6be958c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70de7b5208da4bcd84652f67bbf73fc0 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d9dc6dcdee94767a885a78af2f2f7ae : SUCCESS in 41s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"c271dd8b6cc9d1ece1c5d36077559bec4b3db1b8","revisions":{"2b34b5bf8ebe633474f8704c421f8ba9fe3710fe":{"kind":"REWORK","_number":1,"created":"2023-09-07 06:36:39.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/1"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/2b34b5bf8ebe633474f8704c421f8ba9fe3710fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/2b34b5bf8ebe633474f8704c421f8ba9fe3710fe"}]},"branch":"refs/heads/master"},"f4a0e10f6a554bfea685dbb167e5c08092ee519f":{"kind":"REWORK","_number":2,"created":"2023-09-07 06:51:27.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/2"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:49:36.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/f4a0e10f6a554bfea685dbb167e5c08092ee519f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/f4a0e10f6a554bfea685dbb167e5c08092ee519f"}]},"branch":"refs/heads/master"},"a3a462065a824b86014a75b002e92ea10d505631":{"kind":"REWORK","_number":3,"created":"2023-09-08 09:06:18.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/3"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-08 09:04:18.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a3a462065a824b86014a75b002e92ea10d505631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a3a462065a824b86014a75b002e92ea10d505631"}]},"branch":"refs/heads/master"},"73a3c1760d31a621a8e8323e9ca60d6081856d2e":{"kind":"REWORK","_number":4,"created":"2023-09-08 10:43:40.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/4"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-08 10:41:52.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/73a3c1760d31a621a8e8323e9ca60d6081856d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/73a3c1760d31a621a8e8323e9ca60d6081856d2e"}]},"branch":"refs/heads/master"},"2da0b3d7db3bfa1e3e1b0ce0b1f8ffff8bba4b83":{"kind":"REWORK","_number":5,"created":"2023-09-08 10:59:22.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/5"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-08 10:57:36.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/2da0b3d7db3bfa1e3e1b0ce0b1f8ffff8bba4b83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/2da0b3d7db3bfa1e3e1b0ce0b1f8ffff8bba4b83"}]},"branch":"refs/heads/master"},"11b5f04868c9e6d010b9b5ba336723c0ddd15c52":{"kind":"REWORK","_number":6,"created":"2023-09-11 00:15:24.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/6"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-11 00:13:32.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/11b5f04868c9e6d010b9b5ba336723c0ddd15c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/11b5f04868c9e6d010b9b5ba336723c0ddd15c52"}]},"branch":"refs/heads/master"},"5e995b3287d79fad763ff4a42884f6a86ca92631":{"kind":"REWORK","_number":7,"created":"2023-09-11 00:30:55.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/7"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-11 00:29:07.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/5e995b3287d79fad763ff4a42884f6a86ca92631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/5e995b3287d79fad763ff4a42884f6a86ca92631"}]},"branch":"refs/heads/master"},"ece29813e4c027e355c729bc3dc90e454df3a0f0":{"kind":"REWORK","_number":8,"created":"2023-09-11 01:09:18.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/8"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-11 01:07:31.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/ece29813e4c027e355c729bc3dc90e454df3a0f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/ece29813e4c027e355c729bc3dc90e454df3a0f0"}]},"branch":"refs/heads/master"},"77a54c04695f58a4fe3bf68b511164040cab66a7":{"kind":"NO_CHANGE","_number":9,"created":"2023-09-11 02:56:00.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/9"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-11 02:54:08.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/77a54c04695f58a4fe3bf68b511164040cab66a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/77a54c04695f58a4fe3bf68b511164040cab66a7"}]},"branch":"refs/heads/master"},"75c1c19aa7658a4f2e0a22297480c88e3149001d":{"kind":"REWORK","_number":10,"created":"2023-09-11 05:34:43.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/10"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-11 05:32:51.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/75c1c19aa7658a4f2e0a22297480c88e3149001d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/75c1c19aa7658a4f2e0a22297480c88e3149001d"}]},"branch":"refs/heads/master"},"c55c598f4d6112fb3b1c03a97a44e30f97851f1a":{"kind":"REWORK","_number":11,"created":"2023-10-11 06:15:00.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/11"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-10-11 06:12:47.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c55c598f4d6112fb3b1c03a97a44e30f97851f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c55c598f4d6112fb3b1c03a97a44e30f97851f1a"}]},"branch":"refs/heads/master"},"3cfaf1226adbf93317d352043035ec0a1f8238f9":{"kind":"REWORK","_number":12,"created":"2023-10-11 07:23:54.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/12"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-10-11 07:23:35.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/3cfaf1226adbf93317d352043035ec0a1f8238f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/3cfaf1226adbf93317d352043035ec0a1f8238f9"}]},"branch":"refs/heads/master"},"5c2079418e903be9b26bd66f56d8dc49ddf2436b":{"kind":"REWORK","_number":13,"created":"2023-10-11 08:23:13.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/13"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-10-11 08:22:52.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/5c2079418e903be9b26bd66f56d8dc49ddf2436b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/5c2079418e903be9b26bd66f56d8dc49ddf2436b"}]},"branch":"refs/heads/master"},"6da49151218a0f854725bb2eb486a96679d288f1":{"kind":"REWORK","_number":14,"created":"2023-10-17 01:11:22.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/14"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-10-17 01:10:53.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/6da49151218a0f854725bb2eb486a96679d288f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/6da49151218a0f854725bb2eb486a96679d288f1"}]},"branch":"refs/heads/master"},"f9582c700d89be3892e63bbf655fbbce16e3e71c":{"kind":"REWORK","_number":15,"created":"2023-10-17 01:40:26.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/15"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-10-17 01:39:58.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/f9582c700d89be3892e63bbf655fbbce16e3e71c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/f9582c700d89be3892e63bbf655fbbce16e3e71c"}]},"branch":"refs/heads/master"},"fdab2deb93d433438bfe01d913baa9674c307b7c":{"kind":"REWORK","_number":16,"created":"2023-10-17 02:04:18.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/16"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-10-17 02:03:58.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/fdab2deb93d433438bfe01d913baa9674c307b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/fdab2deb93d433438bfe01d913baa9674c307b7c"}]},"branch":"refs/heads/master"},"c271dd8b6cc9d1ece1c5d36077559bec4b3db1b8":{"kind":"REWORK","_number":17,"created":"2023-10-17 02:49:37.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/05/894005/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/05/894005/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/05/894005/17"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-09-07 06:34:52.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2023-10-17 02:49:19.000000000","tz":480},"subject":"Fixes the several problems/errors caused by tox","message":"Fixes the several problems/errors caused by tox\n\nChange-Id: I75d312ed6f1b9a71cc7cd2817c2dc60d3baa2971\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c271dd8b6cc9d1ece1c5d36077559bec4b3db1b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/c271dd8b6cc9d1ece1c5d36077559bec4b3db1b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
