)]}'
{"id":"openstack%2Fsahara-plugin-cdh~894986","triplet_id":"openstack%2Fsahara-plugin-cdh~master~I6e5b1ce980480b6b7120e4ad7df4578d77e6fcdf","project":"openstack/sahara-plugin-cdh","branch":"master","topic":"release-health-check-cwr-bobcat","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-11-07 08:56:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6e5b1ce980480b6b7120e4ad7df4578d77e6fcdf","subject":"DNM: gate health test","status":"ABANDONED","created":"2023-09-13 19:24:20.000000000","updated":"2023-11-07 08:56:10.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1a63a7b2623d82832ae877f729e6c158fb3f1c60","_number":894986,"virtual_id_number":894986,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-02 07:54:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-11-02 08:03:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-11-02 08:03:08.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-13 19:35:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-02 07:34:10.000000000","updated_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"reviewer":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"state":"CC"},{"updated":"2023-11-02 08:03:08.000000000","updated_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"reviewer":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"state":"REVIEWER"}],"messages":[{"id":"a4e9667a6fa61e0a1a898db416a94d99bb14e492","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-13 19:24:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4057688007fcda88987ae29ab6e2b29a1ba8621a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 19:35:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/368e795cca02436ca752e8f236a2e857\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a18590e6fb4da6b4fb7f791b6044a8 : FAILURE in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48aa1d3dad9645868ee8672c0d21118c : FAILURE in 1m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7148e815d37440197b45bcf2c4b8c5c : FAILURE in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f97719702f547008665ab4ec5738a23 : FAILURE in 2m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/add2a616a46a47cc8da286896e04b7bc : FAILURE in 3m 57s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/f181ae5f28184a16be8e138fc2395735 : FAILURE in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6afd34ca097b80ee898c496861650c8b0c62706f","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-11-02 07:34:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e8e1b7a5c707127ea3418dd57f486d3cf51df18a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 07:54:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8077d2f2dad417485bbfff290e9646e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b57fbdf0cc4a9387cbf1f8291624f8 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c45ceb1402f437e8cded7c2cff8f1cd : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/926d8e21ce214b52b8111eb47a2b3998 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37c0533589cf4c4392199407b032c6bc : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d5df1ef6cc4a70bab0272a4cc72eb9 : SUCCESS in 3m 57s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/bd2238ed0820468d9efb1e2764f3e742 : FAILURE in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41a8f4cc6118c9ecf429abdd1101be92bd94687c","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-11-02 08:03:08.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a63a7b2623d82832ae877f729e6c158fb3f1c60","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-07 08:56:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b27c668068d083bf9f9732bc94c84f106ee43995","revisions":{"b27c668068d083bf9f9732bc94c84f106ee43995":{"kind":"REWORK","_number":1,"created":"2023-09-13 19:24:20.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/86/894986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/86/894986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/86/894986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/86/894986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/86/894986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/86/894986/1"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:24:08.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:24:18.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I6e5b1ce980480b6b7120e4ad7df4578d77e6fcdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/b27c668068d083bf9f9732bc94c84f106ee43995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/b27c668068d083bf9f9732bc94c84f106ee43995"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
