)]}'
{"id":"openstack%2Fsahara-plugin-cdh~895447","triplet_id":"openstack%2Fsahara-plugin-cdh~stable%2F2023.2~Icb3eaccaa21bed1aa93797b8a9ca9b030f1e3a92","project":"openstack/sahara-plugin-cdh","branch":"stable/2023.2","topic":"create-2023.2","hashtags":[],"change_id":"Icb3eaccaa21bed1aa93797b8a9ca9b030f1e3a92","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","status":"MERGED","created":"2023-09-15 14:27:32.000000000","updated":"2023-11-03 06:09:33.000000000","submitted":"2023-11-03 06:07:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895447-create-2023.2","meta_rev_id":"00c1b51f26d8b084a1c804a897567c89d2e1e292","_number":895447,"virtual_id_number":895447,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-03 06:07:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-03 05:55:56.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-03 05:55:56.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-15 14:48:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-03 05:37:15.000000000","updated_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"reviewer":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"state":"CC"},{"updated":"2023-11-03 05:55:56.000000000","updated_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"reviewer":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"state":"REVIEWER"}],"messages":[{"id":"e880cd2f26f8f60e1a662baf6841063679f7f2d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-09-15 14:27:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7be319d83fb10335e59ea01552d12ddb7aac3f30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:48:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b78bd1a1d5b14a9e8b3608423da3c244\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1156fba2db7d419ca4d2842f01541fc4 : FAILURE in 2m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/708ff431a65d4c98ba855a4c7592d350 : FAILURE in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c03e6a5d09c4d7eb79064367f6433a0 : FAILURE in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2da9eee21ccc481aae07b183025489ad : FAILURE in 2m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad00cb10fec8427898984b6ca2da1bde : FAILURE in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9482ca19635d451b9ae97f3c201a99bf : SUCCESS in 4m 17s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/f6cd76b9302440d1a0d443e6f64389fb : FAILURE in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77e27b8488559548cdf349e9d58f74b3f7d55869","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-11-03 05:37:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"083f0e017c33d8aa2d06dde542a15ec8552d8fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 05:48:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/823f1720bd1d41f198d0c1201a0641ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1489651366eb41b7948c934d42e56215 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce23511a158b419db3cf39fe6121fa23 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fe800580e4a4182845158b681db7e50 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d3e9ce8254e432f8eb6dbb6364d3517 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8775f67b485a4976834ad7a8a3c23e89 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c4b84d513484310995f00574c33dfd8 : SUCCESS in 3m 27s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/16dcc076f685471e81fbb55c01cfb50e : FAILURE in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a21675a730ae569ca3bab8d2f9baaddeefc6086","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-11-03 05:55:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"238051e9a3c318bafa2cf8dccdc7df4b5ac872af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 05:56:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1947c3bfe1679eb43f34624a52a2ccbb84836626","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 06:07:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55d3e08d22424965b0775596a7adb0bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba0df378f8c843e4af6bfbbe73d44e10 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/934651d325384cdcaf85119e87c9e5ae : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c334a4466b9482eaa2ce2b10ff27be8 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a5bcd0d29944328adfe5bdcb838b77 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fad7580eec34e7b9ddc3c9ba51e2859 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"128c75e50137aba871dd4b90db5796ebb06f4b8f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 06:07:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"00c1b51f26d8b084a1c804a897567c89d2e1e292","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 06:09:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fcbf313e14140cb8183075baaeb6cea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5a3582ba3fa4f28a25898aeb19bcc74 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1e0634016e546c58a9b60b889fcdbd7 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a0fea1a9495db7686053a30bd00c94a8a3b545b","revisions":{"3a0fea1a9495db7686053a30bd00c94a8a3b545b":{"kind":"REWORK","_number":1,"created":"2023-09-15 14:27:32.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/47/895447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/47/895447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/47/895447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/47/895447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/47/895447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/47/895447/1"}}},"commit":{"parents":[{"commit":"7b5ae3aa315e41c57ed4e366056e28567cdded12","subject":"Update .gitreview for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/7b5ae3aa315e41c57ed4e366056e28567cdded12"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 14:27:31.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 14:27:31.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2023.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2023.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Icb3eaccaa21bed1aa93797b8a9ca9b030f1e3a92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/3a0fea1a9495db7686053a30bd00c94a8a3b545b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/3a0fea1a9495db7686053a30bd00c94a8a3b545b"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
