)]}'
{"id":"openstack%2Fsahara-plugin-cdh~899994","triplet_id":"openstack%2Fsahara-plugin-cdh~stable%2F2023.1~Ifb02067bc2cab9e518bdcc0a2093175caf14f22b","project":"openstack/sahara-plugin-cdh","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"36486":{"account":{"_account_id":36486,"name":"w","display_name":"zhiqiang05","email":"wangzhiqiang05@inspur.com","username":"wangzhiqiang05","status":"programer"},"last_update":"2023-11-03 03:15:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifb02067bc2cab9e518bdcc0a2093175caf14f22b","subject":"fix tox.ini","status":"MERGED","created":"2023-11-03 00:54:16.000000000","updated":"2023-11-03 03:22:13.000000000","submitted":"2023-11-03 03:15:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899994","meta_rev_id":"d9095dd62a2025b9f0039f2ac676b4ebdc923731","_number":899994,"virtual_id_number":899994,"owner":{"_account_id":36486,"name":"w","display_name":"zhiqiang05","email":"wangzhiqiang05@inspur.com","username":"wangzhiqiang05","status":"programer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-03 03:15:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-03 02:18:28.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-03 02:18:28.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-03 01:05:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-03 02:18:28.000000000","updated_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"reviewer":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"state":"REVIEWER"}],"messages":[{"id":"df87fbcd21bfe29a170e0264f0d1c101770095a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36486,"name":"w","display_name":"zhiqiang05","email":"wangzhiqiang05@inspur.com","username":"wangzhiqiang05","status":"programer"},"date":"2023-11-03 00:54:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb7273b61247f34bf3f6ad6212c722bc001c9d35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 01:05:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5db5c2c8e9d84f00b1dcd886d7d91739\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bbe0291cd3943bca2342e003eaefb7e : FAILURE in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f6db0c61f5543e48faf091808bc691f : FAILURE in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4c8c36f4b244b51b3a2ac69416c5feb : FAILURE in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a15a1e710ba949fdbf7564ba7da1c2b6 : FAILURE in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5257b1cce9e04617b0564d077719f761 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/562b85dfc7a541849710400e2a24cce9 : SUCCESS in 2m 51s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/1d311d96bf964b2089224946764f2646 : FAILURE in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8222bbc1e2d2d68e384a990ec9fabcfc1b2e8625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36486,"name":"w","display_name":"zhiqiang05","email":"wangzhiqiang05@inspur.com","username":"wangzhiqiang05","status":"programer"},"date":"2023-11-03 01:51:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"178088518c1e47ba310cafb8f0d24bf5f7acb9c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36486,"name":"w","display_name":"zhiqiang05","email":"wangzhiqiang05@inspur.com","username":"wangzhiqiang05","status":"programer"},"date":"2023-11-03 02:00:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6474a03857a4175efb9d276d299eb605ef172be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 02:17:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d7ff5180e53467baa4f66c46d1c58f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66ece8efec284848bf4df03b4f76c0e8 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c3c5f8ac9894ead8368cc7926ec8d30 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d8f72cd61aa46708ae98de10a138ad2 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76626ae0666d44daac0d5e535c86876e : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e006e85ea04eb99aed5fb32ad807a4 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d49105cbbd54b0c90d811f614ed536b : SUCCESS in 3m 37s\n- sahara-buildimages-cloudera https://zuul.opendev.org/t/openstack/build/444987827f104fd58d3434a027d2421d : FAILURE in 5m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eebeeb8c027130108088d69ea9705802c2af184d","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-11-03 02:18:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"29d838778dd2f415857135743d769b69b933d5f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 02:19:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5d6081ec6cd0b5ee4577a5db94122a85954f5aab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 03:15:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a094ab2d1d2240e3aed4b411b27ece9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56943b3c008c490c8e991f3b1238dd51 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97fb01ed70784dc48dd9a9fd74de695f : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/188a69326a6b4a26a8ad11589fd799a4 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520b0235d3de465ea84a8883697e61bc : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/118f3ba967fb4bff87ee5af8d0f66838 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":3},{"id":"e56c71af4be1ccbde2797f2d7611146aca51f1f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 03:15:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d9095dd62a2025b9f0039f2ac676b4ebdc923731","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 03:22:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cb1dcb4ad9f44e7a4f458750918e465\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7db2f3fe504240e7997f538fe6f2cca1 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0eb42f9b0b1432bbeaab7aaf18ebefa : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ea42e119d4e12db6ce212286c640d818d9ae738c","revisions":{"a2b0446b8eefc93b963f776144840c535d211e23":{"kind":"REWORK","_number":1,"created":"2023-11-03 00:54:16.000000000","uploader":{"_account_id":36486,"name":"w","display_name":"zhiqiang05","email":"wangzhiqiang05@inspur.com","username":"wangzhiqiang05","status":"programer"},"ref":"refs/changes/94/899994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/94/899994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/94/899994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/94/899994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/94/899994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/94/899994/1"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"wangzhiqiang05","email":"wangzhiqiang05@inspur.com","date":"2023-11-03 00:52:49.000000000","tz":480},"committer":{"name":"wangzhiqiang05","email":"wangzhiqiang05@inspur.com","date":"2023-11-03 00:54:08.000000000","tz":480},"subject":"fix tox.ini","message":"fix tox.ini\n\nChange-Id: Ifb02067bc2cab9e518bdcc0a2093175caf14f22b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a2b0446b8eefc93b963f776144840c535d211e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a2b0446b8eefc93b963f776144840c535d211e23"}]},"branch":"refs/heads/stable/2023.1"},"971bf42e2d8a83218183077edf463e084018e603":{"kind":"REWORK","_number":2,"created":"2023-11-03 01:51:03.000000000","uploader":{"_account_id":36486,"name":"w","display_name":"zhiqiang05","email":"wangzhiqiang05@inspur.com","username":"wangzhiqiang05","status":"programer"},"ref":"refs/changes/94/899994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/94/899994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/94/899994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/94/899994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/94/899994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/94/899994/2"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"wangzhiqiang05","email":"wangzhiqiang05@inspur.com","date":"2023-11-03 00:52:49.000000000","tz":480},"committer":{"name":"wangzhiqiang05","email":"wangzhiqiang05@inspur.com","date":"2023-11-03 01:50:23.000000000","tz":480},"subject":"fix tox.ini","message":"fix tox.ini\n\nChange-Id: Ifb02067bc2cab9e518bdcc0a2093175caf14f22b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/971bf42e2d8a83218183077edf463e084018e603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/971bf42e2d8a83218183077edf463e084018e603"}]},"branch":"refs/heads/stable/2023.1"},"ea42e119d4e12db6ce212286c640d818d9ae738c":{"kind":"REWORK","_number":3,"created":"2023-11-03 02:00:04.000000000","uploader":{"_account_id":36486,"name":"w","display_name":"zhiqiang05","email":"wangzhiqiang05@inspur.com","username":"wangzhiqiang05","status":"programer"},"ref":"refs/changes/94/899994/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-cdh","ref":"refs/changes/94/899994/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/94/899994/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/94/899994/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/94/899994/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-cdh refs/changes/94/899994/3"}}},"commit":{"parents":[{"commit":"a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/a888d6be6b50adc346f3c0d9cba7b7ab8274ba8a"}]}],"author":{"name":"wangzhiqiang05","email":"wangzhiqiang05@inspur.com","date":"2023-11-03 00:52:49.000000000","tz":480},"committer":{"name":"wangzhiqiang05","email":"wangzhiqiang05@inspur.com","date":"2023-11-03 01:59:28.000000000","tz":480},"subject":"fix tox.ini","message":"fix tox.ini\n\nChange-Id: Ifb02067bc2cab9e518bdcc0a2093175caf14f22b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/ea42e119d4e12db6ce212286c640d818d9ae738c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-cdh/commit/ea42e119d4e12db6ce212286c640d818d9ae738c"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
