)]}'
{"id":"openstack%2Fsahara-plugin-mapr~667307","triplet_id":"openstack%2Fsahara-plugin-mapr~master~I7a378c69f3ac35ec398400ff8be8421b619a8e6b","project":"openstack/sahara-plugin-mapr","branch":"master","topic":"sphinx-reqs","hashtags":[],"change_id":"I7a378c69f3ac35ec398400ff8be8421b619a8e6b","subject":"Update sphinx from current requirements","status":"MERGED","created":"2019-06-25 09:47:15.000000000","updated":"2019-06-28 17:17:48.000000000","submitted":"2019-06-28 17:17:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667307-1561742268663-be057232","meta_rev_id":"808b818f401ae5b81941c4c3ba8966dc1c8dc562","_number":667307,"virtual_id_number":667307,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":2,"date":"2019-06-28 17:17:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2019-06-28 17:02:20.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":2,"date":"2019-06-25 17:58:15.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":1,"date":"2019-06-28 17:02:20.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 17:58:15.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-06-28 17:02:20.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-06-28 17:17:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0254e36a610f1c868c4cfaa44eae1751ab43f1a6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-06-25 09:47:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d12e76ba2ae8cca65dbc9f218fd757b135adc82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 10:32:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/07/667307/1/check/requirements-check/99d3215/ : SUCCESS in 3m 06s\n- openstack-tox-pep8 http://logs.openstack.org/07/667307/1/check/openstack-tox-pep8/0acd86d/ : SUCCESS in 8m 04s\n- openstack-tox-docs http://logs.openstack.org/07/667307/1/check/openstack-tox-docs/40345ea/html/ : SUCCESS in 3m 45s\n- build-openstack-releasenotes http://logs.openstack.org/07/667307/1/check/build-openstack-releasenotes/bb9a8c4/ : POST_FAILURE in 33m 04s\n- sahara-buildimages-mapr http://logs.openstack.org/07/667307/1/check/sahara-buildimages-mapr/d8020ed/ : FAILURE in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"722ae3a958fd5b47fc643ad8b98830e61cc59d9e","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-06-25 17:35:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ebd1995a2313129da570b719c26df95fc38ed124","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 17:56:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/07/667307/1/check/requirements-check/7e67539/ : SUCCESS in 2m 08s\n- openstack-tox-pep8 http://logs.openstack.org/07/667307/1/check/openstack-tox-pep8/6d1e8eb/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/07/667307/1/check/openstack-tox-docs/819534d/html/ : SUCCESS in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/07/667307/1/check/build-openstack-releasenotes/0aec2c7/html/ : SUCCESS in 3m 06s\n- sahara-buildimages-mapr http://logs.openstack.org/07/667307/1/check/sahara-buildimages-mapr/89a2d82/ : FAILURE in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3925bc194d73435dd465896071ac8bf455b1a3ce","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-06-25 17:58:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"16befefed8e68104913ace9c5dc5e48291933622","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-06-28 17:02:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"84097d3ae700992d61cc5a77dd08fccd3e26c2e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 17:02:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92a825df1b64e4816eae73ac636f2f00ef0ac561","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 17:17:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b13b752eaf6cec784a4c2388be98948967f50a0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 17:17:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/07/667307/1/gate/requirements-check/d532f98/ : SUCCESS in 3m 30s\n- openstack-tox-pep8 http://logs.openstack.org/07/667307/1/gate/openstack-tox-pep8/ab947ef/ : SUCCESS in 8m 26s\n- openstack-tox-docs http://logs.openstack.org/07/667307/1/gate/openstack-tox-docs/b1de22b/html/ : SUCCESS in 4m 48s\n- build-openstack-releasenotes http://logs.openstack.org/07/667307/1/gate/build-openstack-releasenotes/848bab3/html/ : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"495691b42d620f3fb33b087b026886ff3ef04410","revisions":{"495691b42d620f3fb33b087b026886ff3ef04410":{"kind":"REWORK","_number":1,"created":"2019-06-25 09:47:15.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/07/667307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/07/667307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/07/667307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/07/667307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/07/667307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/07/667307/1"}}},"commit":{"parents":[{"commit":"4d2a7e502890d1275776dbc07d4a0086471d7da8","subject":"Update Python 3 test runtimes for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/4d2a7e502890d1275776dbc07d4a0086471d7da8"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-06-25 09:43:09.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-06-25 09:46:35.000000000","tz":120},"subject":"Update sphinx from current requirements","message":"Update sphinx from current requirements\n\nOtherwise the requirements-check job fails.\n\nChange-Id: I7a378c69f3ac35ec398400ff8be8421b619a8e6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/495691b42d620f3fb33b087b026886ff3ef04410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/495691b42d620f3fb33b087b026886ff3ef04410"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
