)]}'
{"id":"openstack%2Fsahara-plugin-mapr~669494","triplet_id":"openstack%2Fsahara-plugin-mapr~master~I7a630e290fa49436f95a9ec2ebc9fcaa7124e5ac","project":"openstack/sahara-plugin-mapr","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I7a630e290fa49436f95a9ec2ebc9fcaa7124e5ac","subject":"Limit envlist to py37 for Python 3 Train goal","status":"MERGED","created":"2019-07-05 20:02:21.000000000","updated":"2019-07-23 14:17:08.000000000","submitted":"2019-07-23 14:17:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669494-1563891428841-cc061412","meta_rev_id":"49cc9e31c29ff773ddae630251f6d6c7e997eb66","_number":669494,"virtual_id_number":669494,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":2,"date":"2019-07-23 14:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"all":[{"value":2,"date":"2019-07-23 13:57:18.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-07-18 13:58:22.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"all":[{"value":1,"date":"2019-07-23 13:57:18.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-13 18:41:29.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-07-14 02:37:43.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-07-14 02:44:51.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-07-18 13:58:22.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-07-23 13:57:18.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-07-23 14:17:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c81b09ab1de2ec40896c9c34c8aa317366b634a9","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 20:02:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de25fdeafbce401c1f258112f3590df602511a39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:41:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/94/669494/1/check/openstack-tox-lower-constraints/99dce66/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/94/669494/1/check/openstack-tox-pep8/c87fd14/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/94/669494/1/check/openstack-tox-py27/e00f336/ : SUCCESS in 6m 09s\n- openstack-tox-py36 http://logs.openstack.org/94/669494/1/check/openstack-tox-py36/27b30d7/ : SUCCESS in 5m 03s\n- openstack-tox-py37 http://logs.openstack.org/94/669494/1/check/openstack-tox-py37/937df11/ : SUCCESS in 5m 06s\n- openstack-tox-docs http://logs.openstack.org/94/669494/1/check/openstack-tox-docs/6456c32/html/ : SUCCESS in 4m 07s\n- build-openstack-releasenotes http://logs.openstack.org/94/669494/1/check/build-openstack-releasenotes/83eab02/html/ : SUCCESS in 3m 21s\n- sahara-buildimages-mapr http://logs.openstack.org/94/669494/1/check/sahara-buildimages-mapr/1fe71e3/ : SUCCESS in 1h 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b889a25cc8702c39f3e84f5882d0811f1119af7","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-07-08 09:24:26.000000000","message":"Patch Set 1: Code-Review-1\n\nIs this expected to remove py36? The only reference to Python 3.6 in the spec seems to hint that Python 3.6 can stay: https://governance.openstack.org/tc/goals/train/python3-updates.html","accounts_in_message":[],"_revision_number":1},{"id":"89afd8449bd34c365ede3e7834fde4dc0b7ddebf","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-08 15:58:29.000000000","message":"Patch Set 1:\n\nHi Luigi, Please see the following for the reply: https://review.opendev.org/669496.","accounts_in_message":[],"_revision_number":1},{"id":"e930710876dab1ecfb1ef64761863b8d2f37bc0c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-07-09 09:34:39.000000000","message":"Patch Set 1: Code-Review+2\n\nAck.","accounts_in_message":[],"_revision_number":1},{"id":"4c8d0b7d6ce9c96cc6beb4436f597b4adc6f63f2","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-13 18:41:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"53fc02aecc4ec93100d435052d07f88245f2c2d9","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-07-14 02:37:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"16ef0826bf5a30762a7bccfab69e804359b59e62","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-14 02:44:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ac0b8d3f6a452e277b764d427b1cf60aa7d09a93","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-07-15 22:08:15.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"170982b084d80460d7a1594b5c0182e04b6271e9","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-07-18 13:57:38.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a0a8fba0f992e1fa6e00f5d5457187f1427571bf","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-07-18 13:58:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f13d19e9eab2f99e840dc4b424dc2ac95c34bfea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 15:31:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/94/669494/2/check/openstack-tox-lower-constraints/0bcf73a/ : SUCCESS in 3m 10s\n- openstack-tox-pep8 http://logs.openstack.org/94/669494/2/check/openstack-tox-pep8/ad7fa06/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/94/669494/2/check/openstack-tox-py27/05ffa9d/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/94/669494/2/check/openstack-tox-py36/c17ab89/ : SUCCESS in 4m 14s\n- openstack-tox-py37 http://logs.openstack.org/94/669494/2/check/openstack-tox-py37/fab8523/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/94/669494/2/check/openstack-tox-docs/880c65d/html/ : SUCCESS in 3m 42s\n- build-openstack-releasenotes http://logs.openstack.org/94/669494/2/check/build-openstack-releasenotes/c4d24ec/html/ : SUCCESS in 3m 18s\n- sahara-buildimages-mapr http://logs.openstack.org/94/669494/2/check/sahara-buildimages-mapr/5eb2acd/ : SUCCESS in 1h 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6098591885a1caa85d33a5a5a1760eb2583d2b2f","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-07-23 13:57:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"113325f97f299ab9811668dcb34bda9427769972","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 13:57:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c533bae3de98c613aed05e8eb308cab210df60a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:17:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/94/669494/2/gate/openstack-tox-lower-constraints/9e61cc0/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/94/669494/2/gate/openstack-tox-pep8/1f8d050/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/94/669494/2/gate/openstack-tox-py27/41f39a5/ : SUCCESS in 5m 01s\n- openstack-tox-py36 http://logs.openstack.org/94/669494/2/gate/openstack-tox-py36/738ffe4/ : SUCCESS in 4m 32s\n- openstack-tox-py37 http://logs.openstack.org/94/669494/2/gate/openstack-tox-py37/6a39a12/ : SUCCESS in 5m 11s\n- openstack-tox-docs http://logs.openstack.org/94/669494/2/gate/openstack-tox-docs/7492c1e/html/ : SUCCESS in 3m 15s\n- build-openstack-releasenotes http://logs.openstack.org/94/669494/2/gate/build-openstack-releasenotes/59d0f86/html/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"e9ed298b62c0f6cbb5d94545303567850d07eb0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:17:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"317af9ee159989b8b7711dccd0fc2aafaaa99f53","revisions":{"b681a440c977232f4c4e7cc4687ff0a5a9b1bd43":{"kind":"REWORK","_number":1,"created":"2019-07-05 20:02:21.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/94/669494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/94/669494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/94/669494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/94/669494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/94/669494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/94/669494/1"}}},"commit":{"parents":[{"commit":"495691b42d620f3fb33b087b026886ff3ef04410","subject":"Update sphinx from current requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/495691b42d620f3fb33b087b026886ff3ef04410"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 16:00:33.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 20:02:21.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I7a630e290fa49436f95a9ec2ebc9fcaa7124e5ac\nStory: #2005924\nTask: #34244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/b681a440c977232f4c4e7cc4687ff0a5a9b1bd43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/b681a440c977232f4c4e7cc4687ff0a5a9b1bd43"}]},"branch":"refs/heads/master"},"317af9ee159989b8b7711dccd0fc2aafaaa99f53":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-18 13:57:38.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/94/669494/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/94/669494/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/94/669494/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/94/669494/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/94/669494/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/94/669494/2"}}},"commit":{"parents":[{"commit":"495691b42d620f3fb33b087b026886ff3ef04410","subject":"Update sphinx from current requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/495691b42d620f3fb33b087b026886ff3ef04410"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 16:00:33.000000000","tz":-240},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-07-18 13:56:39.000000000","tz":0},"subject":"Limit envlist to py37 for Python 3 Train goal","message":"Limit envlist to py37 for Python 3 Train goal\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train. The correct zuul jobs are\nalready in place so this change just limits default \u0027tox\u0027 calls to py37\nto save developers time.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I7a630e290fa49436f95a9ec2ebc9fcaa7124e5ac\nStory: #2005924\nTask: #34244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/317af9ee159989b8b7711dccd0fc2aafaaa99f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/317af9ee159989b8b7711dccd0fc2aafaaa99f53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
