)]}'
{"id":"openstack%2Fsahara-plugin-mapr~676018","triplet_id":"openstack%2Fsahara-plugin-mapr~stable%2Fstein~I20eeac76373058eca0c33da11eb1e9dfce91de54","project":"openstack/sahara-plugin-mapr","branch":"stable/stein","hashtags":[],"change_id":"I20eeac76373058eca0c33da11eb1e9dfce91de54","subject":"Check MariaDB installation","status":"MERGED","created":"2019-08-12 20:22:24.000000000","updated":"2019-08-28 18:52:26.000000000","submitted":"2019-08-28 18:48:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676018-1567018092859-91f773fe","meta_rev_id":"3ad82f959ee4257811b20c476d06351f4f6cca65","_number":676018,"virtual_id_number":676018,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":2,"date":"2019-08-28 18:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2019-08-28 18:37:55.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":1,"date":"2019-08-28 18:37:55.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-28 18:37:55.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-08-28 18:48:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3404d50ad4faec4c4a06220454c41e925f176c6e","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-08-12 20:22:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb6d404b69fdb0907c49afd7a8e6c3bd12e81370","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 20:39:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/648fe025ad124779b5a9fbeb2d033fd3 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e523456436646e4bada9f14426e26db : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/545a9438264e4b6d8b12a69ce7dcfd12 : SUCCESS in 3m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5d4e422b7a7343088c4ae68cd1cf4791 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d5e21344fa643c6b6d0d0ac6de38703 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a58cc24d4fd04e5791db2ec7f84879ca : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38521f24c85e46eb886680e1232df67a : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f51b412ae648b9cf9da7271b76f2f231a02d91da","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-08-28 18:37:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ebccf2fa307ece60e23da2e5ab7a9a3ac628b759","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 18:38:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0627810ed55326c21eaeabac3ed54bd1f858f598","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 18:48:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6db00dff898d495abe79a0e6458a4835 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/669052dbd3264d97ae0e20e0d830bd0a : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e3dd44dc20a414d86482b61abd08dd3 : SUCCESS in 6m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5a7b1fa42da84fdca47f3178d92b9369 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cf01e4e96794d9b8e2074681b1f32d4 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/733b2991d74041d781257b327234092e : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dee932c49ec4f88990a28f7d43d5438 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b3cc319837110b5c1c2c8b8fab618b0cef5303d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 18:48:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3ad82f959ee4257811b20c476d06351f4f6cca65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 18:52:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34b9f7635fc14e61b1d7149d0571345d : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f92ebb03437f76697489f4df8331c451623dc52","revisions":{"1f92ebb03437f76697489f4df8331c451623dc52":{"kind":"REWORK","_number":1,"created":"2019-08-12 20:22:24.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/18/676018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/18/676018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/18/676018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/18/676018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/18/676018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/18/676018/1"}}},"commit":{"parents":[{"commit":"7f4e060dd35743681f51b181a16955dac04c042e","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/7f4e060dd35743681f51b181a16955dac04c042e"}]}],"author":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-07-26 18:44:04.000000000","tz":-180},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-08-12 20:22:04.000000000","tz":120},"subject":"Check MariaDB installation","message":"Check MariaDB installation\n\nChecking if MariaDB is installed before trying to install it.\n\nStory: #2005902\nTask: #33764\n\nChange-Id: I20eeac76373058eca0c33da11eb1e9dfce91de54\n(cherry picked from commit 7df4957949fed406b708b850677a4a41d7926690)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/1f92ebb03437f76697489f4df8331c451623dc52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/1f92ebb03437f76697489f4df8331c451623dc52"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
