)]}'
{"id":"openstack%2Fsahara-plugin-mapr~692925","triplet_id":"openstack%2Fsahara-plugin-mapr~stable%2Ftrain~I23c8d14dce45d003fdfb58fed6591e937b27abab","project":"openstack/sahara-plugin-mapr","branch":"stable/train","topic":"fix_image_creation","hashtags":[],"change_id":"I23c8d14dce45d003fdfb58fed6591e937b27abab","subject":"Fixing image creation","status":"MERGED","created":"2019-11-04 23:45:45.000000000","updated":"2019-11-06 13:28:44.000000000","submitted":"2019-11-06 13:27:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692925-1573046855852-c91a571c","meta_rev_id":"aaf6f94e99977c137029064f3a87592e2ce7f85a","_number":692925,"virtual_id_number":692925,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-06 13:27:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2019-11-06 13:14:18.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":2,"date":"2019-11-05 00:24:55.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":1,"date":"2019-11-06 13:14:18.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 00:24:55.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-11-06 13:14:18.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-11-06 13:27:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2421b43ffa9076383271faf1355f99c4566f655","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-04 23:45:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c44ee589a2bd6d8fdc191a796b3a1c9b7d825b4c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-04 23:49:12.000000000","message":"Topic set to fix_image_creation","accounts_in_message":[],"_revision_number":1},{"id":"d77ce8697c9372979da1b9dc86b69b4be2ac54ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 00:04:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bc408bfe7dc4550a33182c806b47257 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e8c69cabfd437ea997d9eda06d0d74 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59443da619ba40d9853813ad4f76a73d : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f6928a9b2da49e796c2d6c35792f501 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/736f907f5f7447c6b4e891f96b9f7cdd : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7040639255424235a41ad4be158d3c33 : SUCCESS in 6m 12s\n- sahara-buildimages-mapr https://zuul.opendev.org/t/openstack/build/86349e7f881840689ec340697d20cee1 : SUCCESS in 17m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25f6e5bab4ed6277ff09f1ee56d2b03d6da8b593","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-11-05 00:24:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"628f9379351dc29582d1ac155200ed9a99974b4d","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-11-06 13:14:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"26da1113128410b58d9890cfd4279ebc806f8dba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 13:14:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3911caa51fbc729b8d97b1318fa0f7da6b0b513b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 13:27:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"def127cd68c470b7cf62e81e6c8a95050cd63c4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 13:27:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48daf588f2d9400da48fa186b30e3128 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74118815303640ef80f86982aac842f4 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95ba0b87c82343b78a9c1fb183d87f67 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fc985a915054f84882c18425080c187 : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52de19fdbdb443b0b2ce3ceb4dc95049 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/295ab36faccc482299f28c37a01c5965 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":1},{"id":"aaf6f94e99977c137029064f3a87592e2ce7f85a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 13:28:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4722fb7598d347ec8962d0920e04138e : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7b37aac861c515aa073d02fe34f06e497ca4acb","revisions":{"a7b37aac861c515aa073d02fe34f06e497ca4acb":{"kind":"REWORK","_number":1,"created":"2019-11-04 23:45:45.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/25/692925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/25/692925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/25/692925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/25/692925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/25/692925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/25/692925/1"}}},"commit":{"parents":[{"commit":"f682f9aab5631c24b968f8c68df119ecce27f8eb","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/f682f9aab5631c24b968f8c68df119ecce27f8eb"}]}],"author":{"name":"Telles Nobrega","email":"tellesnobrega@gmail.com","date":"2019-10-29 17:25:26.000000000","tz":-180},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-11-04 23:45:35.000000000","tz":60},"subject":"Fixing image creation","message":"Fixing image creation\n\nUpdating image creation to keep compatibility with some package changes.\n\nStory: #2006752\nTask: #37236\nChange-Id: I23c8d14dce45d003fdfb58fed6591e937b27abab\n(cherry picked from commit 2cf6806c1829bf1ac0915f78ec319f07967f254b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/a7b37aac861c515aa073d02fe34f06e497ca4acb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/a7b37aac861c515aa073d02fe34f06e497ca4acb"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
