)]}'
{"id":"openstack%2Fsahara-plugin-mapr~722097","triplet_id":"openstack%2Fsahara-plugin-mapr~master~Ief8e8cf4bc7a85621189f5e050e3b60d0cb59de0","project":"openstack/sahara-plugin-mapr","branch":"master","topic":"project-ptl-and-contrib-docs","hashtags":[],"change_id":"Ief8e8cf4bc7a85621189f5e050e3b60d0cb59de0","subject":"Ussuri contributor docs community goal","status":"MERGED","created":"2020-04-22 17:44:25.000000000","updated":"2020-04-23 11:41:38.000000000","submitted":"2020-04-23 11:40:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722097-1587642003351-fbe8d8ff","meta_rev_id":"dcabd67fc78085bb355e2a26a06bb78ea23297fd","_number":722097,"virtual_id_number":722097,"owner":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 11:40:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2020-04-23 11:25:18.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":2,"date":"2020-04-22 18:29:09.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":1,"date":"2020-04-23 11:25:18.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 18:29:09.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-04-23 11:25:18.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2020-04-23 11:40:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6293c100a7cf54e485a5c1402971eaf83b187ca","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2020-04-22 17:44:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"002c74379615c033ce65284197ca918c73dc31f7","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2020-04-22 17:50:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e49bb53f8aba2ee0de2e428cdd84fabd7afda66c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 18:08:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4cf8841e6c949bfb864f736549229a6 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/101398694d6841139984c85f175f6bb5 : SUCCESS in 10m 09s\n- sahara-buildimages-mapr https://zuul.opendev.org/t/openstack/build/93292adabe1a4b09a2e3d70bd11d711c : FAILURE in 5m 35s (non-voting)\n\nWarning:\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/install_java.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/images/common/resources/package_utils.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/add_mapr_repo.sh","accounts_in_message":[],"_revision_number":2},{"id":"f49fb86eaca0f8ff04f9f3db9a684bd9c4c21da0","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-22 18:29:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b1d21b351015e9082a2c672b597953cf754f3f6d","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2020-04-23 11:25:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a922f231a6758bf72e363358f2b95e21ff10f0fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:25:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ca5c383eb5389fbae3999674349b6d6bfca6ce0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:40:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/384a4e16858b45439da082ac52ffbec7 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fa72f4bae3f4739bdd2fa2f61249d0c : SUCCESS in 7m 24s\n\nWarning:\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/install_java.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/images/common/resources/package_utils.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/add_mapr_repo.sh","accounts_in_message":[],"_revision_number":2},{"id":"957eba76682cbabc743893e506ed501eeafcd5cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:40:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cf3460f2cedd95cfc0044e3ef116d681c6cb7ea2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:41:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b320c92bbc384fbb85b492cd0887b0df : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"498cc2998d70f3137150d31cd6acd022946de450","revisions":{"a31bf6863c8b9253554d2f995b70699b0e11c151":{"kind":"REWORK","_number":1,"created":"2020-04-22 17:44:25.000000000","uploader":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"ref":"refs/changes/97/722097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/97/722097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/97/722097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/97/722097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/97/722097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/97/722097/1"}}},"commit":{"parents":[{"commit":"61ee828da2aa92139def8bd5db88139250356bbc","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/61ee828da2aa92139def8bd5db88139250356bbc"}]}],"author":{"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","date":"2020-04-22 17:43:30.000000000","tz":-240},"committer":{"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","date":"2020-04-22 17:43:30.000000000","tz":-240},"subject":"Ussuri contributor docs community goal","message":"Ussuri contributor docs community goal\n\n- Updated CONTRIBUTING.rst per goal guidelines\n- Created doc/source/contributor/contributing.rst per goal guidelines\n\nChange-Id: Ief8e8cf4bc7a85621189f5e050e3b60d0cb59de0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/a31bf6863c8b9253554d2f995b70699b0e11c151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/a31bf6863c8b9253554d2f995b70699b0e11c151"}]},"branch":"refs/heads/master"},"498cc2998d70f3137150d31cd6acd022946de450":{"kind":"REWORK","_number":2,"created":"2020-04-22 17:50:51.000000000","uploader":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"ref":"refs/changes/97/722097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/97/722097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/97/722097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/97/722097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/97/722097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/97/722097/2"}}},"commit":{"parents":[{"commit":"61ee828da2aa92139def8bd5db88139250356bbc","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/61ee828da2aa92139def8bd5db88139250356bbc"}]}],"author":{"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","date":"2020-04-22 17:43:30.000000000","tz":-240},"committer":{"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","date":"2020-04-22 17:50:18.000000000","tz":-240},"subject":"Ussuri contributor docs community goal","message":"Ussuri contributor docs community goal\n\n- Updated CONTRIBUTING.rst per goal guidelines\n- Created doc/source/contributor/contributing.rst per goal guidelines\n\nChange-Id: Ief8e8cf4bc7a85621189f5e050e3b60d0cb59de0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/498cc2998d70f3137150d31cd6acd022946de450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/498cc2998d70f3137150d31cd6acd022946de450"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
