)]}'
{"id":"openstack%2Fsahara-plugin-mapr~722440","triplet_id":"openstack%2Fsahara-plugin-mapr~stable%2Fussuri~Iabaa866ad2d60b99558ad74ab991bb8765f62d28","project":"openstack/sahara-plugin-mapr","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"Iabaa866ad2d60b99558ad74ab991bb8765f62d28","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-23 19:12:47.000000000","updated":"2020-04-23 22:03:56.000000000","submitted":"2020-04-23 22:01:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722440-1587679309853-5004bf0a","meta_rev_id":"a74b90289ceb2be69ca04d1a2f1d6c8af5cdb663","_number":722440,"virtual_id_number":722440,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 22:01:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":2,"date":"2020-04-23 21:47:18.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":1,"date":"2020-04-23 21:47:18.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 21:47:18.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-04-23 22:01:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63a1aa39268ad9567e0a152c2ff63284b3b551e0","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-23 19:12:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a853d783009972e9f54f5066746e1d476f3b6f17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 19:42:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d27d9c6f400f4b7e9a6085772a9d7268 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e77ff27dc354cb2afdf858569357fe0 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02c73c34c9f94ffca5ff2d3c86454660 : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70ef300501ca40c190d5b6bcbfdb91c0 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19c6b3e31be4472da09389b83347179c : SUCCESS in 7m 45s\n- sahara-buildimages-mapr https://zuul.opendev.org/t/openstack/build/b204671e69274b79bfc642ce072b9c0f : FAILURE in 3m 51s (non-voting)\n\nWarning:\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/install_java.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/images/common/resources/package_utils.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/add_mapr_repo.sh","accounts_in_message":[],"_revision_number":1},{"id":"01fa1c1aa0138164670df22f59c20e014e66f2c5","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-23 21:47:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee3a33c05184b7eaa5a0a82fb5983086f388fdca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 21:47:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"971af7a28535297072a0fa4c9048d2717e5eb121","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 22:01:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d811cfd24424c7eab15bed4ef1b8af8 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add5bae846c94aae9e70097da667a66c : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c72d2c5d9c9940a88e11e33d2fda0b8d : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fa99216acf84520934dc7f19c7ff504 : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce7c107640e04c4ab0d503856006c0dc : SUCCESS in 8m 54s\n\nWarning:\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/install_java.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/images/common/resources/package_utils.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/add_mapr_repo.sh","accounts_in_message":[],"_revision_number":1},{"id":"7c24ba259f34a27057d4d62f2682ba5203e93478","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 22:01:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dffab8d864a44b7732704a65e612b2a469a3fc43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 22:03:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/239fc7d62934410992898f6b1930ef1d : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37cc2739781b4d43b10e08903f2fa639d5679714","revisions":{"37cc2739781b4d43b10e08903f2fa639d5679714":{"kind":"REWORK","_number":1,"created":"2020-04-23 19:12:47.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/40/722440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/40/722440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/40/722440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/40/722440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/40/722440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/40/722440/1"}}},"commit":{"parents":[{"commit":"498cc2998d70f3137150d31cd6acd022946de450","subject":"Ussuri contributor docs community goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/498cc2998d70f3137150d31cd6acd022946de450"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 19:12:45.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 19:12:45.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: Iabaa866ad2d60b99558ad74ab991bb8765f62d28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/37cc2739781b4d43b10e08903f2fa639d5679714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/37cc2739781b4d43b10e08903f2fa639d5679714"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
