)]}'
{"id":"openstack%2Fsahara-plugin-mapr~722901","triplet_id":"openstack%2Fsahara-plugin-mapr~master~Icf32770e0beedc78ba4cf1c29cc371105be01faf","project":"openstack/sahara-plugin-mapr","branch":"master","topic":"py38-tox","hashtags":[],"change_id":"Icf32770e0beedc78ba4cf1c29cc371105be01faf","subject":"Bump default tox env from py37 to py38","status":"MERGED","created":"2020-04-24 15:36:43.000000000","updated":"2020-04-24 23:36:53.000000000","submitted":"2020-04-24 23:32:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722901-1587771153298-16b58591","meta_rev_id":"d96ef2eaa26d79b2cabb37ef1729f7f14109a43c","_number":722901,"virtual_id_number":722901,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 23:32:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":2,"date":"2020-04-24 23:17:01.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":1,"date":"2020-04-24 23:17:01.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 23:17:01.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-04-24 23:32:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc4534a45e6e9b75dbbdb6fe7bffd261bf56751a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 15:36:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9b4086ce81ab7ad31bd3c3f052cac6130cbdff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 16:38:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86362396206d46b1b33f2422bd45237d : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c96411dfca794c98b39c19a91522d222 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94af499e3c0b41f48a10abc8b63ea278 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42dbd1e5666046fcbc499343c40a4f5f : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380b2a7508ee41ff8219d33ba2544edf : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9bc6c26a6df4d5598db184fbd7991c6 : SUCCESS in 7m 31s\n- sahara-buildimages-mapr https://zuul.opendev.org/t/openstack/build/1bad7d537d2b4b6fa5a01ac621364741 : FAILURE in 5m 22s (non-voting)\n\nWarning:\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/install_java.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/images/common/resources/package_utils.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/add_mapr_repo.sh","accounts_in_message":[],"_revision_number":1},{"id":"002ee18d8fcb639b5d09b62ffdb7b7ab373f7474","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-24 23:17:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7eebb0a1d9fee854a03c9c10399149fb3030d6a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 23:17:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"687428d6450a726ab1d0e4b698f268d1e27697a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 23:32:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d22bccb352ed4358b667d8e7dde8b13d : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8e60d6a5d5742b9b4810d6c1f0bf987 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2ad83fb4e0848f29c6f3380a966581a : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25b4c5e71d754771870a6f827e6d2a11 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be153bb230cb49c3bdc8bd12fae1a3cf : SUCCESS in 12m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/959622251c07465dbdb82dc551c68f46 : SUCCESS in 8m 04s\n\nWarning:\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/install_java.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/images/common/resources/package_utils.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/add_mapr_repo.sh","accounts_in_message":[],"_revision_number":1},{"id":"38f631fe02474c91890ef8b84442e4df463da49a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 23:32:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"02db02dd6ab251b0c82202d2b44f85e471a0ac1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 23:36:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf0469d6fa1c48a0aea50863750bd14f : SUCCESS in 3m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f12bca55b7ac46f49cca70764d60e736 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6d3eda9fd3ad6aadf45eeb911a8ea3782f5653c","revisions":{"d6d3eda9fd3ad6aadf45eeb911a8ea3782f5653c":{"kind":"REWORK","_number":1,"created":"2020-04-24 15:36:43.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/01/722901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/01/722901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/01/722901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/01/722901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/01/722901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/01/722901/1"}}},"commit":{"parents":[{"commit":"eab33b6b94f8d5c4d1f44994b2e066786a9d785d","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/eab33b6b94f8d5c4d1f44994b2e066786a9d785d"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:26:01.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:26:01.000000000","tz":-300},"subject":"Bump default tox env from py37 to py38","message":"Bump default tox env from py37 to py38\n\nPython 3.8 is now our highest level supported python runtime.\nThis updates the default tox target environments to swap out\npy37 for py38 to make sure local development testing is\ncovering this version.\n\nThis does not impact zuul jobs in any way, nor prevent local\ntests against py37. It just changes the default if none is\nexplicitly provided.\n\nChange-Id: Icf32770e0beedc78ba4cf1c29cc371105be01faf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/d6d3eda9fd3ad6aadf45eeb911a8ea3782f5653c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/d6d3eda9fd3ad6aadf45eeb911a8ea3782f5653c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
