)]}'
{"id":"openstack%2Fsahara-plugin-mapr~850841","triplet_id":"openstack%2Fsahara-plugin-mapr~master~I242c200a5c403eea35d21e4e8c25bcb23bc756f9","project":"openstack/sahara-plugin-mapr","branch":"master","hashtags":[],"change_id":"I242c200a5c403eea35d21e4e8c25bcb23bc756f9","subject":"Use TOX_CONSTRAINTS_FILE","status":"ABANDONED","created":"2022-07-25 08:12:10.000000000","updated":"2024-05-15 05:07:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c5d5448c4b2cd085ab38c1acf26b24000ed9a5d9","_number":850841,"virtual_id_number":850841,"owner":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-25 08:26:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 08:26:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5b2c320fe02d33010d08499b97dcaa2507f53e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-07-25 08:12:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"233a4098d8a4f66870dafc3fdb282c865561f670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 08:26:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68cf3cefc8624c8dae9abdd5f06c4b6e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/781ae53835c04936b1f4ef66a3f5ca4f : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9c388f7f5fb45248b8f61eb2ea53e4b : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccc94374bf52411f898a436a5b608052 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d07987331a68432f80fa78bde7cc98e7 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a902449aef436ebfb099e13643d24e : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/860cf009a5f14755a17e972f660172b9 : SUCCESS in 3m 31s\n- sahara-buildimages-mapr https://zuul.opendev.org/t/openstack/build/987b1eb3542e4e5ca726e579f8b6d33b : FAILURE in 5m 41s (non-voting)\n\nWarning:\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/add_mapr_repo.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/images/common/resources/package_utils.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/install_java.sh","accounts_in_message":[],"_revision_number":1},{"id":"c5d5448c4b2cd085ab38c1acf26b24000ed9a5d9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-15 05:07:16.000000000","message":"Abandoned\n\nsahara project has been retired https://review.opendev.org/c/openstack/governance/+/919374","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc45f82bbc4b1bd8617d53ba47127b7567bf6ea7","revisions":{"cc45f82bbc4b1bd8617d53ba47127b7567bf6ea7":{"kind":"REWORK","_number":1,"created":"2022-07-25 08:12:10.000000000","uploader":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"ref":"refs/changes/41/850841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/41/850841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/41/850841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/41/850841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/41/850841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/41/850841/1"}}},"commit":{"parents":[{"commit":"3708548d0ba92a46c04638a1ea34510be03c2f05","subject":"Delete py36 and py37","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/3708548d0ba92a46c04638a1ea34510be03c2f05"}]}],"author":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-07-25 08:11:01.000000000","tz":480},"committer":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-07-25 08:11:13.000000000","tz":480},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I242c200a5c403eea35d21e4e8c25bcb23bc756f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/cc45f82bbc4b1bd8617d53ba47127b7567bf6ea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/cc45f82bbc4b1bd8617d53ba47127b7567bf6ea7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
