)]}'
{"id":"openstack%2Fsahara-plugin-mapr~874900","triplet_id":"openstack%2Fsahara-plugin-mapr~master~Iba1a68e29ec7371cf559f8f12e8ddecfb7a2260d","project":"openstack/sahara-plugin-mapr","branch":"master","topic":"release-health-check-cwr-antelope","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-11-07 08:59:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iba1a68e29ec7371cf559f8f12e8ddecfb7a2260d","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2023-02-23 10:49:55.000000000","updated":"2023-11-07 08:59:40.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"39430cbaa4de2642e653152d2d3022289d13c4ec","_number":874900,"virtual_id_number":874900,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-02 01:58:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 11:01:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-02 01:47:57.000000000","updated_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"reviewer":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"state":"CC"}],"messages":[{"id":"05adb3a20b8e969955bd5cb2ff70a03536989778","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-23 10:49:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a426790a5f4beed5cd73f21cd0b2271c7fcfae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 11:01:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1af9c3ff4134286973efd1606b32b0e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35dcc06e8f7843a9bc6a83e3c3f8a345 : FAILURE in 2m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc53680da601417ca3a5053f0af23929 : FAILURE in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6182edff6bb14171a84c1ae3f91ccb67 : FAILURE in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d70cae340e6d4537a8acb0db3a486280 : FAILURE in 4m 02s\n- sahara-buildimages-mapr https://zuul.opendev.org/t/openstack/build/6f36c52623cd43f5b32ebc66468604be : FAILURE in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ac09fa28bc0f9fe789b4a89e7ace8ed75764aba","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-11-02 01:47:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3e6641fa1ab9e1382503dabe00facf87d68d8a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 01:58:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bc5601199244c98b2fac5b135bca6c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/069501e61d8d4591b1443aa3c4dacc24 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e651b57f0a64c998477b67d30a4675c : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4af28155305b464b8f657c761f91c158 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07b1a0faef242799deb6f611156e88c : SUCCESS in 4m 38s\n- sahara-buildimages-mapr https://zuul.opendev.org/t/openstack/build/c3536b76582b42f6a7f230b418e2f712 : FAILURE in 4m 38s (non-voting)\n\nWarning:\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/add_mapr_repo.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/images/common/resources/package_utils.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/install_java.sh","accounts_in_message":[],"_revision_number":1},{"id":"39430cbaa4de2642e653152d2d3022289d13c4ec","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-07 08:59:40.000000000","message":"Abandoned\n\nGate is not blocking, abandoning.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"28c7588c59ec9bc6381547d37ce0f4cc8bd570fd","revisions":{"28c7588c59ec9bc6381547d37ce0f4cc8bd570fd":{"kind":"REWORK","_number":1,"created":"2023-02-23 10:49:55.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/00/874900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/00/874900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/00/874900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/00/874900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/00/874900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/00/874900/1"}}},"commit":{"parents":[{"commit":"e745b5f47c9c6a02b80581d1ea69a46723330b31","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/e745b5f47c9c6a02b80581d1ea69a46723330b31"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-23 10:49:45.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-23 10:49:53.000000000","tz":60},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: Iba1a68e29ec7371cf559f8f12e8ddecfb7a2260d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/28c7588c59ec9bc6381547d37ce0f4cc8bd570fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/28c7588c59ec9bc6381547d37ce0f4cc8bd570fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
