)]}'
{"id":"openstack%2Fsahara-plugin-mapr~894987","triplet_id":"openstack%2Fsahara-plugin-mapr~master~I6dacf7bb6f02541faebe13576463aa999177dbb5","project":"openstack/sahara-plugin-mapr","branch":"master","topic":"release-health-check-cwr-bobcat","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-11-06 14:39:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6dacf7bb6f02541faebe13576463aa999177dbb5","subject":"DNM: gate health test","status":"ABANDONED","created":"2023-09-13 19:24:36.000000000","updated":"2023-11-06 14:39:37.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9a7089ee0e02e61f9f429b4211869fd1063791c5","_number":894987,"virtual_id_number":894987,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-02 01:58:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-13 19:35:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-02 01:47:28.000000000","updated_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"reviewer":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"state":"CC"}],"messages":[{"id":"4851ff50665d5e40a7d223cde8bdb4789e5c65a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-13 19:24:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5eb19051800e3df43d6885ecebfbc88d71f197b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 19:35:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce05f593cf234fb38c194217dcfc92b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca61d46951504ad594bc7f2c4f86dbad : FAILURE in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62b9aa29ce584d6f9254636eec29f722 : FAILURE in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78648d1f1d824182a3de378ff79a16df : FAILURE in 2m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c01dfb19e5c4f4385d0c89e661473a4 : FAILURE in 2m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5b42570afcc4f3db641182dfec7dde2 : FAILURE in 2m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30988d15a34d4c2bb9e0eedfc82abda6 : FAILURE in 3m 31s\n- sahara-buildimages-mapr https://zuul.opendev.org/t/openstack/build/0ea10537562a40c680669263cd5f40f3 : FAILURE in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"276758a59abb54c6668bee9cf9868b0cefed917e","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-11-02 01:47:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6479a82272789a364db958325f6ae16dd01c874c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 01:58:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a6c56b8d3f74ad68f8cff1c14bbff85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00ffaa5f17da4c609887b9ca7773b70a : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe23d3ae686a4ed7803abd00dcc18544 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bffba9989a6d457db9261c00526a9cf1 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1430f43e014f82916f5846de0811ab : SUCCESS in 4m 57s\n- sahara-buildimages-mapr https://zuul.opendev.org/t/openstack/build/716c5d51217b47a2a35cc23994f48b19 : FAILURE in 4m 33s (non-voting)\n\nWarning:\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/add_mapr_repo.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/images/common/resources/package_utils.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/install_java.sh","accounts_in_message":[],"_revision_number":1},{"id":"9a7089ee0e02e61f9f429b4211869fd1063791c5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-11-06 14:39:37.000000000","message":"Abandoned\n\nGate is fixed (thanks!). Abandoning.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96620bca00d375e02c9a29327505c1596d39ed8f","revisions":{"96620bca00d375e02c9a29327505c1596d39ed8f":{"kind":"REWORK","_number":1,"created":"2023-09-13 19:24:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/87/894987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/87/894987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/87/894987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/87/894987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/87/894987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/87/894987/1"}}},"commit":{"parents":[{"commit":"e745b5f47c9c6a02b80581d1ea69a46723330b31","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/e745b5f47c9c6a02b80581d1ea69a46723330b31"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:24:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-09-13 19:24:34.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I6dacf7bb6f02541faebe13576463aa999177dbb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/96620bca00d375e02c9a29327505c1596d39ed8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/96620bca00d375e02c9a29327505c1596d39ed8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
