)]}'
{"id":"openstack%2Fsahara-plugin-mapr~895456","triplet_id":"openstack%2Fsahara-plugin-mapr~master~I31caeb2c5951026f22a0d9a87c3724712c325c3f","project":"openstack/sahara-plugin-mapr","branch":"master","topic":"reno-2023.2","hashtags":[],"change_id":"I31caeb2c5951026f22a0d9a87c3724712c325c3f","subject":"Update master for stable/2023.2","status":"MERGED","created":"2023-09-15 14:30:21.000000000","updated":"2023-11-02 01:56:50.000000000","submitted":"2023-11-02 01:55:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895456-reno-2023.2","meta_rev_id":"30c5941b5335f1b6f72de0b847b51a67732a6676","_number":895456,"virtual_id_number":895456,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-02 01:55:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-02 01:46:17.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-02 01:46:17.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-15 14:53:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-02 01:33:20.000000000","updated_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"reviewer":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"state":"CC"},{"updated":"2023-11-02 01:46:17.000000000","updated_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"reviewer":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"state":"REVIEWER"}],"messages":[{"id":"23e3486c9dcfb16c86169822795a0875b34515e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-09-15 14:30:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4329d4c3f10a26986681dcf2842bdcf430a68775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:53:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f70088304714ee781d2332d85fd1c15\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9c350d2d504ba9a535a5a4cfe382d4 : FAILURE in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7bd5d91e49b431386c334d1f205e5fe : FAILURE in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51df3323c6114fc1aba90acc0f1671c1 : SUCCESS in 4m 54s\n- sahara-buildimages-mapr https://zuul.opendev.org/t/openstack/build/ec85233220234726bfc0b51349482b8b : FAILURE in 5m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb51a47e0d88b22ebec5ad34db2a5bc9601f03ef","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-11-02 01:33:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab79dd0e1ce2ce39c2bdf34d098df7a37c014faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 01:42:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfdc7755effb475db0f2b506154c10a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9b52a400c134f01b5941a031d4fe210 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e9e819cfb0401c82d2dda979aaae8f : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84c262a7bc474607893ca3b44c12de4e : SUCCESS in 3m 14s\n- sahara-buildimages-mapr https://zuul.opendev.org/t/openstack/build/d947824a39034caea6b4d5276d5fc8fa : FAILURE in 4m 47s (non-voting)\n\nWarning:\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/add_mapr_repo.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/images/common/resources/package_utils.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/install_java.sh","accounts_in_message":[],"_revision_number":1},{"id":"dd26c05c1f74b757299c61da712b6f0e2db63499","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-11-02 01:46:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"17973ecbdfb5eb2833b98867a42aca7f8eabf7a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 01:46:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4368c5b1f877d4b608ef56093b9a14277913746f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 01:55:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16420cc6e69c427caf2f052d6acea929\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5246191a0ac242e39febe351d7d7e5b8 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2887b00c9f4d488e0a3bb5df72f2d4 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc53e96994bf4aae9086fc248abfe724 : SUCCESS in 3m 00s\n\nWarning:\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/add_mapr_repo.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/images/common/resources/package_utils.sh\n  Comments left for invalid file sahara_plugin_mapr/plugins/mapr/resources/install_java.sh","accounts_in_message":[],"_revision_number":1},{"id":"478147931b04d2b3b08091fcc579ffbbe9c11e7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 01:55:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"30c5941b5335f1b6f72de0b847b51a67732a6676","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 01:56:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/355d2d7392cb4a48932f95ddab9d30e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeaf862e032b496593243b0ad8afd901 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10016eca368443fab0b3387772f515aa : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e1a4e6dd166ffc6d75d9c73f1577be43b4a2d83","revisions":{"0e1a4e6dd166ffc6d75d9c73f1577be43b4a2d83":{"kind":"REWORK","_number":1,"created":"2023-09-15 14:30:21.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/56/895456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-mapr","ref":"refs/changes/56/895456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/56/895456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/56/895456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/56/895456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-mapr refs/changes/56/895456/1"}}},"commit":{"parents":[{"commit":"e745b5f47c9c6a02b80581d1ea69a46723330b31","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/e745b5f47c9c6a02b80581d1ea69a46723330b31"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 14:30:20.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-15 14:30:20.000000000","tz":0},"subject":"Update master for stable/2023.2","message":"Update master for stable/2023.2\n\nAdd file to the reno documentation build to show release notes for\nstable/2023.2.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/2023.2.\n\nSem-Ver: feature\nChange-Id: I31caeb2c5951026f22a0d9a87c3724712c325c3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/0e1a4e6dd166ffc6d75d9c73f1577be43b4a2d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-mapr/commit/0e1a4e6dd166ffc6d75d9c73f1577be43b4a2d83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
