)]}'
{"id":"openstack%2Fsahara-plugin-spark~654563","triplet_id":"openstack%2Fsahara-plugin-spark~stable%2Fstein~I06ef32cfb156bb1d6a3aef42ceb441193cacc290","project":"openstack/sahara-plugin-spark","branch":"stable/stein","hashtags":[],"change_id":"I06ef32cfb156bb1d6a3aef42ceb441193cacc290","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"ABANDONED","created":"2019-04-22 16:39:52.000000000","updated":"2019-04-25 14:40:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f2b192f86913421f5f5c75dbf8dcc9a68525f9bf","_number":654563,"virtual_id_number":654563,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-04-22 17:49:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 17:49:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"314f633a4b6661bf0a43a88743f996461ed88df7","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-22 16:39:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71d37a56d16d44d503d1a86ca7b85d2fb9a21543","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:49:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/63/654563/1/check/test-release-openstack/13dddf4/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/63/654563/1/check/openstack-tox-pep8/1831571/ : SUCCESS in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/63/654563/1/check/openstack-tox-docs/12dcaa2/html/ : SUCCESS in 5m 19s\n- sahara-buildimages-spark http://logs.openstack.org/63/654563/1/check/sahara-buildimages-spark/cc21394/ : SUCCESS in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2b192f86913421f5f5c75dbf8dcc9a68525f9bf","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-04-25 14:40:42.000000000","message":"Abandoned\n\nDuring today\u0027s Sahara meeting, and following -infra suggestion, it was deemed not necessary to backport those replacements on the older branches.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"514d89baea5f26fe2854f7b4b6fc0090577a72a3","revisions":{"514d89baea5f26fe2854f7b4b6fc0090577a72a3":{"kind":"REWORK","_number":1,"created":"2019-04-22 16:39:52.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/63/654563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-spark","ref":"refs/changes/63/654563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-spark refs/changes/63/654563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-spark refs/changes/63/654563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-spark refs/changes/63/654563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-spark refs/changes/63/654563/1"}}},"commit":{"parents":[{"commit":"d6128c0a861e24141638f4b13a0f1f84939131e5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-spark/commit/d6128c0a861e24141638f4b13a0f1f84939131e5"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-18 13:12:15.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-04-22 16:38:41.000000000","tz":120},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org\ncounterparts.\n\nChange-Id: I06ef32cfb156bb1d6a3aef42ceb441193cacc290\n(cherry picked from commit c30fd5f2e169d48110a4075cd4115334b767ad0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-spark/commit/514d89baea5f26fe2854f7b4b6fc0090577a72a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-spark/commit/514d89baea5f26fe2854f7b4b6fc0090577a72a3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
