)]}'
{"id":"openstack%2Fsahara-plugin-spark~684657","triplet_id":"openstack%2Fsahara-plugin-spark~master~Ia63c17676168278b9a03eec4077f96af3b80bbc6","project":"openstack/sahara-plugin-spark","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ia63c17676168278b9a03eec4077f96af3b80bbc6","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-09-25 09:21:33.000000000","updated":"2019-09-25 10:00:03.000000000","submitted":"2019-09-25 09:58:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684657-1569405500068-91403195","meta_rev_id":"11b96e0f0ba3059d6252bc555f8fbe88d833c05c","_number":684657,"virtual_id_number":684657,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 09:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":2,"date":"2019-09-25 09:43:21.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":1,"date":"2019-09-25 09:43:21.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 09:43:21.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-09-25 09:58:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b5237628657bcad619d2464e707a6761a564aa6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-25 09:21:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce32032665639b3f6c7b80a15804a2ec24aad48b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-09-25 09:43:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4e57ebff76a22e77d858939f15b2824ce58731f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 09:48:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a743bf9a1e7b4b158a224aa3a1cd0a18 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93db35009d1493881c400a9655123be : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f668317308a41ed8a54bab8a3bab4ed : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc8a94b1e2994a868e76338e92b93240 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1e9b8405a1740d19a23e2898e2c352c : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3275a2247a84698b6ea517073bef639 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3339ad264c194c58a380816183cbbf27 : SUCCESS in 5m 12s\n- sahara-buildimages-spark https://zuul.opendev.org/t/openstack/build/2d4e93eafd4949ec8d088e5766c3ce6f : SUCCESS in 12m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccd3ceff1b9fc3d25ffdaebe88e9fe87918bb020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 09:48:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b79c0f38d5d45379aff0cc638f936d33c56cd50a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 09:58:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9c61a669a5442c5ba3c1d4384242ad5 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baaf4803cf6f4ab3afeb04980919c8fb : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6dcaa60485a4f1fa3e15babc6d67fe1 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22ea33cd7c3b463daf3f5573c5a3b6c5 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c49901e1cf64a48a2d3ec7f8ca03565 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f440ce30b9954b9283d6f4dc3f88fd1e : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81e82859aa0c49a99a3f06500901141f : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fe9d7ae8d1851916a1abfb5d683d7bc7b2bd242a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 09:58:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2d1599b64abdcd4e0313f8ae873d30fd0732258c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 10:00:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82e95135470042f4b966e857b17698f9 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0765cd4d65f4006b524e24d9ff9e773 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bfecb1efbc4fd16b343118cc34f046b91842123","revisions":{"4bfecb1efbc4fd16b343118cc34f046b91842123":{"kind":"REWORK","_number":1,"created":"2019-09-25 09:21:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/684657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-spark","ref":"refs/changes/57/684657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-spark refs/changes/57/684657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-spark refs/changes/57/684657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-spark refs/changes/57/684657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-spark refs/changes/57/684657/1"}}},"commit":{"parents":[{"commit":"46453c73ec3102578b06e321ad640ac928e6645c","subject":"Doc updates: bump theme to 1.20.0, add PDF build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-spark/commit/46453c73ec3102578b06e321ad640ac928e6645c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-25 09:21:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-25 09:21:31.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ia63c17676168278b9a03eec4077f96af3b80bbc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-spark/commit/4bfecb1efbc4fd16b343118cc34f046b91842123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-spark/commit/4bfecb1efbc4fd16b343118cc34f046b91842123"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
