)]}'
{"id":"openstack%2Fsahara-plugin-storm~657928","triplet_id":"openstack%2Fsahara-plugin-storm~master~Ib92186e88a3a54e0b6b00a4480fe6b461ad2f372","project":"openstack/sahara-plugin-storm","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Ib92186e88a3a54e0b6b00a4480fe6b461ad2f372","subject":"Update Python 3 test runtimes for Train","status":"MERGED","created":"2019-05-09 06:26:43.000000000","updated":"2019-05-16 17:53:08.000000000","submitted":"2019-05-16 17:53:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657928-1558029188834-a663f06a","meta_rev_id":"7b9621a44044d448a58c6c45330c4ddc95eae622","_number":657928,"virtual_id_number":657928,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":2,"date":"2019-05-16 17:53:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2019-05-09 12:32:39.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":1,"date":"2019-05-10 11:14:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-05-16 16:56:47.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-05-16 16:56:47.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 12:32:39.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-05-10 11:14:12.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-05-16 16:56:47.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-05-16 17:53:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a1ebd203311f9330ddf77a9b35cd65c892ef2be","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-09 06:26:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bc2e1fc49f580fcfc99982acd4dc1d46db7b095","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 06:41:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/28/657928/1/check/test-release-openstack/300e49b/ : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/28/657928/1/check/openstack-tox-lower-constraints/95896f9/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/28/657928/1/check/openstack-tox-pep8/a2af6c4/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/28/657928/1/check/openstack-tox-py27/5ee5bdd/ : SUCCESS in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/28/657928/1/check/openstack-tox-py36/b827a60/ : SUCCESS in 4m 47s\n- openstack-tox-py37 http://logs.openstack.org/28/657928/1/check/openstack-tox-py37/863222e/ : SUCCESS in 4m 58s\n- openstack-tox-docs http://logs.openstack.org/28/657928/1/check/openstack-tox-docs/52e42d9/html/ : SUCCESS in 4m 53s\n- build-openstack-releasenotes http://logs.openstack.org/28/657928/1/check/build-openstack-releasenotes/b9a7478/html/ : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2b55a0e086e10655bb438b0dd85ebdbdd24cf64e","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-09 09:36:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2543b0ca4fadaaaef66fa8e338538445a039ca9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 09:48:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/28/657928/2/check/test-release-openstack/aa6fe9b/ : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/28/657928/2/check/openstack-tox-lower-constraints/4325735/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/28/657928/2/check/openstack-tox-pep8/02f277f/ : SUCCESS in 5m 07s\n- openstack-tox-py27 http://logs.openstack.org/28/657928/2/check/openstack-tox-py27/87da953/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/28/657928/2/check/openstack-tox-py36/8d01788/ : SUCCESS in 4m 54s\n- openstack-tox-py37 http://logs.openstack.org/28/657928/2/check/openstack-tox-py37/fdd4161/ : SUCCESS in 4m 51s\n- openstack-tox-docs http://logs.openstack.org/28/657928/2/check/openstack-tox-docs/dab3a53/html/ : SUCCESS in 5m 25s\n- build-openstack-releasenotes http://logs.openstack.org/28/657928/2/check/build-openstack-releasenotes/5ad534b/html/ : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":2},{"id":"88b8f4d64a2a69b49ad2c5ec25ac52dc5cab551d","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-05-09 12:32:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"14b23a7c22e463d99173dd5883fd6b27522795e4","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-10 11:14:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d48beef674504954850611529aec9a24feb4a3d5","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-05-16 16:56:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9271f1c3053e72adf78e1004593ecfc516b26e33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:57:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"520e4e0f4dd0e55bf8ec3c6b33d0432fc34fcadf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 17:53:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0200a803fd86931c14f199aa8fa44a7486d6433a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 17:53:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/28/657928/2/gate/test-release-openstack/fc3a577/ : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/28/657928/2/gate/openstack-tox-lower-constraints/df55ac9/ : SUCCESS in 5m 36s\n- openstack-tox-pep8 http://logs.openstack.org/28/657928/2/gate/openstack-tox-pep8/cc24e93/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/28/657928/2/gate/openstack-tox-py27/4082a23/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/28/657928/2/gate/openstack-tox-py36/6d0079f/ : SUCCESS in 4m 40s\n- openstack-tox-py37 http://logs.openstack.org/28/657928/2/gate/openstack-tox-py37/47c291c/ : SUCCESS in 5m 03s\n- openstack-tox-docs http://logs.openstack.org/28/657928/2/gate/openstack-tox-docs/73a0ee5/html/ : SUCCESS in 4m 37s\n- build-openstack-releasenotes http://logs.openstack.org/28/657928/2/gate/build-openstack-releasenotes/4777f9a/html/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0416da1912a1faf57a95e5f42d8b16a8aab241f4","revisions":{"1dd6664fd20966b5d8b308ef09fbb2616e5283cf":{"kind":"REWORK","_number":1,"created":"2019-05-09 06:26:43.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/28/657928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-storm","ref":"refs/changes/28/657928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/28/657928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/28/657928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/28/657928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/28/657928/1"}}},"commit":{"parents":[{"commit":"464127461fed6b2c2bbc2ea32bb72310021d1c0c","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-storm/commit/464127461fed6b2c2bbc2ea32bb72310021d1c0c"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:00.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:00.000000000","tz":480},"subject":"Update Python 3 test runtimes for Train","message":"Update Python 3 test runtimes for Train\n\nFollowing the process set down in the TC resolution on updating Python\nversions:\n\nhttps://governance.openstack.org/tc/resolutions/20181024-python-update-process.html\n\nChange-Id: Ib92186e88a3a54e0b6b00a4480fe6b461ad2f372\nDepends-On: https://review.opendev.org/#/c/641878/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-storm/commit/1dd6664fd20966b5d8b308ef09fbb2616e5283cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-storm/commit/1dd6664fd20966b5d8b308ef09fbb2616e5283cf"}]},"branch":"refs/heads/master"},"0416da1912a1faf57a95e5f42d8b16a8aab241f4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-09 09:36:01.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/28/657928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-storm","ref":"refs/changes/28/657928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/28/657928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/28/657928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/28/657928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/28/657928/2"}}},"commit":{"parents":[{"commit":"464127461fed6b2c2bbc2ea32bb72310021d1c0c","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-storm/commit/464127461fed6b2c2bbc2ea32bb72310021d1c0c"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:00.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 09:34:48.000000000","tz":480},"subject":"Update Python 3 test runtimes for Train","message":"Update Python 3 test runtimes for Train\n\nThis goal is to implement the process set out in the 2018-10-24 Python\nUpdate Process TC resolution[1], for the Train cycle to ensure unit\ntesting is in place for all of the Tested Runtimes for Train[2].\nIn practice, this generally means adding unit tests for Python 3.7 and dropping\nunit tests for Python 3.5. Using the Zuul template for Train will ensure that\nall projects that support Python3 will be tested against the agreed runtime\nversions, and make it easier to update them in future.\n\n[1]https://governance.openstack.org/tc/resolutions/20181024-python-update-process.html\n[2]https://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Ib92186e88a3a54e0b6b00a4480fe6b461ad2f372\nDepends-On: https://review.opendev.org/#/c/641878/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-storm/commit/0416da1912a1faf57a95e5f42d8b16a8aab241f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-storm/commit/0416da1912a1faf57a95e5f42d8b16a8aab241f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
