)]}'
{"id":"openstack%2Fsahara-plugin-storm~853030","triplet_id":"openstack%2Fsahara-plugin-storm~master~I272bf6ada4dcbc759fc57f99b078ad1c1880b629","project":"openstack/sahara-plugin-storm","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2023-08-18 02:55:24.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"34665":{"account":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"last_update":"2024-05-15 05:11:29.000000000","reason":"Change was abandoned"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2023-03-01 13:00:02.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}},"hashtags":[],"change_id":"I272bf6ada4dcbc759fc57f99b078ad1c1880b629","subject":"Remove Python 2 support","status":"ABANDONED","created":"2022-08-13 06:10:44.000000000","updated":"2024-05-15 05:11:29.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"97881aa666b4a3e5b63d85fba1f1635ce7d844dc","_number":853030,"virtual_id_number":853030,"owner":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-06 08:10:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2023-03-09 02:20:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"},{"value":1,"date":"2023-03-01 13:00:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-19 01:21:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-19 01:21:20.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-13 06:24:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 07:47:09.000000000","updated_by":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2023-03-09 02:20:32.000000000","updated_by":{"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"},"reviewer":{"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"},"state":"REVIEWER"},{"updated":"2023-05-19 01:21:20.000000000","updated_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"reviewer":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"state":"REVIEWER"},{"updated":"2023-07-06 08:00:44.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"4787a851f9d4ed96703d008f19fcd0e1f7e37395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"date":"2022-08-13 06:10:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f06b54d8ad87765394ff23b471dcc4db813bf16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 06:24:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b729b7a7e1f46bd8ad870509f2eb639\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/dfb124a6c0f54c688d0c49788ac3eac6 : SUCCESS in 3m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63046085890e4ed3a1680fb385538ee8 : FAILURE in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb72c58699f47f0839d8f038732a4e8 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f868bb2df5c847a892b0ece7c9bedfd0 : FAILURE in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b593dc19b394cc9913a77e846835c3d : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ead5e1a521ea489497095794d0edf70f : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d4e7999cbc44a2891461883148182d4 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37bf1f80f5cb426aba8fe198e0b8cd66 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"63e38eae63f44f07179f202bb9877ba46c88f7a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"date":"2022-09-16 05:46:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"654bb71015e7ca5225c2e3c0ea30725262ca5078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 05:56:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0a12d00c8854298bcaa7bd5e13a6a38\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c8e9ae5055fe4fcdbb3386f6f9aa993e : SUCCESS in 1m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0679acf44b204288b24d25f4794b73c2 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d7ba5d32660405fbbfd8a82fb49049a : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51fbfb2eb0c8452fab5c2d3cc6813677 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99515480887145c096204418ac905d80 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e249327e270c4557ba88a1e754d23529 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce44dc76bfe54fa6934ccbdcc30b7690 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":2},{"id":"e28f46ca6191165a0e6ae648659b981fc5896343","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2023-03-01 13:00:02.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4da0311d9c9046b1bc03e052cf4d7686ab4e0ccd","author":{"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"},"date":"2023-03-09 02:20:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"17b4392eb80f87da603ab715a9ae9ee0f66b3728","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-05-19 01:21:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aa8730f9fb289e24016329eb4588cd6fa8f27300","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 01:21:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b7cec3cb9370ad4a5b3465bb6c3f43e415460fd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 01:30:55.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68973ca530f54b24a691a8b2469b0b3b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f8ac82e9724c43c5a1996bacf0b47cdf : SUCCESS in 1m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17d1657d2c0c45d6947d92b1e27be7cf : FAILURE in 1m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/201124903b33451ebe7b041f46228d6c : FAILURE in 2m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1871cf879d340fd86bb5c549311a9c1 : FAILURE in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f19f5396e81446f3ac17f2896dd114eb : FAILURE in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e421ff6d406f40c79e0f0c82a251f84f : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"683d18276afd03ef38c25f6138374f47fa4b0e86","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2023-07-06 08:00:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a1696d33f77809182b44864e2661efe14f0922fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 08:10:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b95cb30c5a874912bf1bc5fab56ba004\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2828b2d4b53345e3ba20076c6d66b90f : SUCCESS in 1m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f1b0c5064614af1a969472e7c07a165 : FAILURE in 2m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/893d8be781e244178f5277954250730c : FAILURE in 1m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2240bd69c6e6469db37be09aff475297 : FAILURE in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2a430bdeca5408cbad8a851aa8a8304 : FAILURE in 2m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35de7ffef2d1446f8416639d34de4b15 : FAILURE in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c0610870892454892e27a1877e9dc3b : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9f491e708aa48092bdd28374c215cc7255560bce","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-08-18 02:55:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"699f97b6b760e5d615e61430a56a7ddc5938ac45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 03:02:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e23d401638c4a2a98d503561a1d3d00\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9f3e58cf3bfd4534a2322bb6fc66b57f : SUCCESS in 1m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75aef6ae3dd342b788af47d5e08d7998 : FAILURE in 2m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bf3f7e4d32441dca8a9c9e89a7c888c : FAILURE in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c2879eea94a496eb33ee83ab7e92ab6 : FAILURE in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c30a235000ea47f495595f3e52528d06 : FAILURE in 2m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00035a270fc348c189341edcb7e8c066 : FAILURE in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c437fcd5f6540848e9fea9af12a24c8 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"97881aa666b4a3e5b63d85fba1f1635ce7d844dc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-15 05:11:29.000000000","message":"Abandoned\n\nsahara project has been retired https://review.opendev.org/c/openstack/governance/+/919374","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"263f7a51753df2815fbb3a94038f0009bd668469","revisions":{"b47bc8314a6e852424fb2276a3437b0dd37523cb":{"kind":"REWORK","_number":1,"created":"2022-08-13 06:10:44.000000000","uploader":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"ref":"refs/changes/30/853030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-storm","ref":"refs/changes/30/853030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/30/853030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/30/853030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/30/853030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/30/853030/1"}}},"commit":{"parents":[{"commit":"ae96409b4001c64e5e07e72e72737b9178a3cce4","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-storm/commit/ae96409b4001c64e5e07e72e72737b9178a3cce4"}]}],"author":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-13 06:09:31.000000000","tz":480},"committer":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-13 06:09:31.000000000","tz":480},"subject":"Remove Python 2 support","message":"Remove Python 2 support\n\nPython 2 has been deprecated. This patch removes all\ntraces of six, unicode strings and Python 2 tweaks.\n\nChange-Id: I272bf6ada4dcbc759fc57f99b078ad1c1880b629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-storm/commit/b47bc8314a6e852424fb2276a3437b0dd37523cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-storm/commit/b47bc8314a6e852424fb2276a3437b0dd37523cb"}]},"branch":"refs/heads/master"},"263f7a51753df2815fbb3a94038f0009bd668469":{"kind":"REWORK","_number":2,"created":"2022-09-16 05:46:04.000000000","uploader":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"ref":"refs/changes/30/853030/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-storm","ref":"refs/changes/30/853030/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/30/853030/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/30/853030/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/30/853030/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-storm refs/changes/30/853030/2"}}},"commit":{"parents":[{"commit":"39af1159ead0d66612ca7b67baa78cd8fa108495","subject":"Dropping lower constraints testing and remove py36,py37 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-storm/commit/39af1159ead0d66612ca7b67baa78cd8fa108495"}]}],"author":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-13 06:09:31.000000000","tz":480},"committer":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-09-16 05:44:41.000000000","tz":480},"subject":"Remove Python 2 support","message":"Remove Python 2 support\n\nPython 2 has been deprecated. This patch removes all\ntraces of six, unicode strings and Python 2 tweaks.\n\nChange-Id: I272bf6ada4dcbc759fc57f99b078ad1c1880b629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-storm/commit/263f7a51753df2815fbb3a94038f0009bd668469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-storm/commit/263f7a51753df2815fbb3a94038f0009bd668469"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
