)]}'
{"id":"openstack%2Fsahara-plugin-vanilla~852986","triplet_id":"openstack%2Fsahara-plugin-vanilla~master~I9d6fa7b57824cc8480dc7f3728d3abfae2b032da","project":"openstack/sahara-plugin-vanilla","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2024-05-15 05:09:51.000000000","reason":"Change was abandoned"},"34665":{"account":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"last_update":"2024-05-15 05:09:51.000000000","reason":"Change was abandoned"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2023-03-08 06:11:34.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}},"hashtags":[],"change_id":"I9d6fa7b57824cc8480dc7f3728d3abfae2b032da","subject":"Remove Python 2 support","status":"ABANDONED","created":"2022-08-13 03:17:06.000000000","updated":"2024-05-15 05:09:51.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4df5771ed1429ed4fded534063c9df1c5bbe1fe2","_number":852986,"virtual_id_number":852986,"owner":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-06 08:12:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"recommended":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2023-03-09 02:19:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"},{"value":1,"date":"2023-03-08 06:11:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-19 01:24:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-19 01:24:54.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-13 03:26:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 07:46:35.000000000","updated_by":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2023-03-09 02:19:20.000000000","updated_by":{"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"},"reviewer":{"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"},"state":"REVIEWER"},{"updated":"2023-05-19 01:24:54.000000000","updated_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"reviewer":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"state":"REVIEWER"},{"updated":"2023-07-06 08:01:03.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"24c301f2dd4d6295baae7f77dc217ad4e1b30b5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"date":"2022-08-13 03:17:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ad21321eb1d6451a3164a0b87e52abbe432d9db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 03:26:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/210520a6d1e6474280bef0be41663c2c\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8af22a32bdbe4159ab6429b4de16dcf4 : FAILURE in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76c27470c0c74bf39a975c81223743f6 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c35e87df30c241e6b32fecb9efb93066 : FAILURE in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb4a3280404c4221b4fc203ccadafd89 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a497588aa1554974acd8a438446ff0db : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06871a9bf0f54ceda2042bb22979b65f : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b540f8e659a34ca4b97822747dbea01f : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ef5320fd713df6b2d3ebaf566eff854b254d28b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"date":"2022-09-16 05:37:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"17237d2f653b4a601087b53e6f1714ab72a79c25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 05:50:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16b990d0e6614dee9ca28759fd2b3847\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ef58e7cf9648fe9219ae03f0ca6aaf : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfa1f8d61da64cd1bb2f55dc441b0ef7 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ad03556fdf1441f851d557413f9af8f : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9148312fcfac4eca80a5ffa1eab51e9c : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f390aecb53d4dec8677e43dbff7ec34 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16e079a5f11b4eb0bf16bad873941a10 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"d98b1c81ab66ec4e83fd9e58146a1b8b5285de58","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2023-03-08 06:11:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"02ced3e0b370e4b91631d4e1ea95cb38d9dad2db","author":{"_account_id":33353,"name":"ruifaling","display_name":"ruifaling","email":"ruifaling@inspur.com","username":"ruifaling"},"date":"2023-03-09 02:19:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a8e1a844c4595e8bc5260f4c4b2ab8bccd6f0333","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-05-19 01:24:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e77ea9fa7d0fb65cb7dd4f59affa3f89124067bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 01:25:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cfed4aa014d515f29e02ee1f2f2ec1dd33bb2a34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 01:41:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a18b9d3241e0422699ee7830c1010131\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5de3fa99caa0497ba5766d3d77b297df : FAILURE in 2m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1916c620d8884a59baab60102a54b6b5 : FAILURE in 1m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a754875753e34496bd26a698681767ef : FAILURE in 2m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae3a87330cd4f0ea491e340770113da : FAILURE in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f40f490d88ff411598e000a525c54983 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"ee7a939b79bdc20afde435715bda39d58e41ab6c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2023-07-06 08:02:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df5c8d584045208d47a3074d58d60c5362cb3478","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 08:12:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ae1fef5dd4f4f0b837b024e702f4b73\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2805fcbbf91b405d827e926a452ac545 : FAILURE in 2m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95ecfb9f759d40ca96a62d7983b0712f : FAILURE in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac0cd4a6b03d418db6f4d6d545d4a447 : FAILURE in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/988e57a847664872b11c6c7ba1509850 : FAILURE in 2m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb93d86411a543a5ad6ea42d3019430a : FAILURE in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d2fc153f18a4ee49c1c19e791a0c099 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":2},{"id":"e500f6b63022f5e1f5ee7831be82cd2af74c4b90","author":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"date":"2023-07-14 01:21:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"093dcc707000c7dc19afc8250cbda0e32cca7fc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 01:29:52.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df1fdaaccdc34d159c86c2cc159e274a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d64384b01cca4f138aa9d31e397d570f : FAILURE in 1m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4395c4a7b394417b5b1839193fc42d1 : FAILURE in 1m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9573c8f39032484fa43f579b7d4a43ca : FAILURE in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/039a8f0dce2f4917a58310d4aa901bf2 : FAILURE in 2m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c70e753a5134417a4bd5803b97f512e : FAILURE in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e21e3b1a639042c999f6185dd3f7be71 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"4df5771ed1429ed4fded534063c9df1c5bbe1fe2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-15 05:09:51.000000000","message":"Abandoned\n\nsahara project has been retired https://review.opendev.org/c/openstack/governance/+/919374","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"500babec2a9ae04542af614ffa5055d665c382d6","revisions":{"4398c9c7a3dd72ffafdebdac254aecd279b2a270":{"kind":"REWORK","_number":1,"created":"2022-08-13 03:17:06.000000000","uploader":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"ref":"refs/changes/86/852986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-vanilla","ref":"refs/changes/86/852986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-vanilla refs/changes/86/852986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-vanilla refs/changes/86/852986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-vanilla refs/changes/86/852986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-vanilla refs/changes/86/852986/1"}}},"commit":{"parents":[{"commit":"6cd493d42097a092ef133338a09835ea6158c7e7","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-vanilla/commit/6cd493d42097a092ef133338a09835ea6158c7e7"}]}],"author":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-13 03:15:15.000000000","tz":480},"committer":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-13 03:15:15.000000000","tz":480},"subject":"Remove Python 2 support","message":"Remove Python 2 support\n\nPython 2 has been deprecated. This patch removes all\ntraces of six, unicode strings and Python 2 tweaks.\n\nChange-Id: I9d6fa7b57824cc8480dc7f3728d3abfae2b032da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-vanilla/commit/4398c9c7a3dd72ffafdebdac254aecd279b2a270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-vanilla/commit/4398c9c7a3dd72ffafdebdac254aecd279b2a270"}]},"branch":"refs/heads/master"},"500babec2a9ae04542af614ffa5055d665c382d6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-16 05:37:59.000000000","uploader":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"ref":"refs/changes/86/852986/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-vanilla","ref":"refs/changes/86/852986/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-vanilla refs/changes/86/852986/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-vanilla refs/changes/86/852986/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-vanilla refs/changes/86/852986/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-vanilla refs/changes/86/852986/2"}}},"commit":{"parents":[{"commit":"d93365f15187ddb0f618317ad6a1d45d700acef5","subject":"Dropping lower constraints testing and remove py36,py37 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-vanilla/commit/d93365f15187ddb0f618317ad6a1d45d700acef5"}]}],"author":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-08-13 03:15:15.000000000","tz":480},"committer":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2022-09-16 05:37:59.000000000","tz":0},"subject":"Remove Python 2 support","message":"Remove Python 2 support\n\nPython 2 has been deprecated. This patch removes all\ntraces of six, unicode strings and Python 2 tweaks.\n\nChange-Id: I9d6fa7b57824cc8480dc7f3728d3abfae2b032da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-vanilla/commit/500babec2a9ae04542af614ffa5055d665c382d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-vanilla/commit/500babec2a9ae04542af614ffa5055d665c382d6"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
