)]}'
{"id":"openstack%2Fsahara-plugin-vanilla~876427","triplet_id":"openstack%2Fsahara-plugin-vanilla~master~I93a92fb9d83ade0121762d62f6c34f992b399599","project":"openstack/sahara-plugin-vanilla","branch":"master","attention_set":{},"removed_from_attention_set":{"11805":{"account":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"last_update":"2024-05-15 05:09:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I93a92fb9d83ade0121762d62f6c34f992b399599","subject":"Drop b1 from min version in requirements.txt","status":"ABANDONED","created":"2023-03-03 19:24:48.000000000","updated":"2024-05-15 05:09:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"478aafffdcf3c919872f4d64e689db0c9eae0f9a","_number":876427,"virtual_id_number":876427,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-03 19:39:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-03 19:39:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e072553002cb4d5e9d3ef99a9fade5683eefcd40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-03-03 19:24:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0712ae57e3b2a368a389395ac03f92191dc941c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 19:39:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/863a0ed5a21f41a58e5153637d66f70f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d9828ed706d543749afc33d7568ddc33 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c48ed59b81b14e4ab0f8f0aba122e7f1 : FAILURE in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54983da6fb734ea58d94caad2ac1cecb : FAILURE in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5149cdaea7b34686b8f1cf51717ccb48 : FAILURE in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/579a2d2da72e4d03b4835463ec376c6d : FAILURE in 3m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/900074c8a94e4ce282e6a7bf9747bc58 : FAILURE in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"478aafffdcf3c919872f4d64e689db0c9eae0f9a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-15 05:09:48.000000000","message":"Abandoned\n\nsahara project has been retired https://review.opendev.org/c/openstack/governance/+/919374","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"937405e6c9131647a8faa4c6b56486c387941929","revisions":{"937405e6c9131647a8faa4c6b56486c387941929":{"kind":"REWORK","_number":1,"created":"2023-03-03 19:24:48.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/27/876427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-plugin-vanilla","ref":"refs/changes/27/876427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-plugin-vanilla refs/changes/27/876427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-plugin-vanilla refs/changes/27/876427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-plugin-vanilla refs/changes/27/876427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-plugin-vanilla refs/changes/27/876427/1"}}},"commit":{"parents":[{"commit":"d93365f15187ddb0f618317ad6a1d45d700acef5","subject":"Dropping lower constraints testing and remove py36,py37 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-vanilla/commit/d93365f15187ddb0f618317ad6a1d45d700acef5"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-03-03 19:24:36.000000000","tz":-300},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-03-03 19:24:47.000000000","tz":-300},"subject":"Drop b1 from min version in requirements.txt","message":"Drop b1 from min version in requirements.txt\n\nThe min versions for sahara has been set to a beta version\nfor a while now. Let\u0027s set it to the official corresponding\nrelease versions.\n\nChange-Id: I93a92fb9d83ade0121762d62f6c34f992b399599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-vanilla/commit/937405e6c9131647a8faa4c6b56486c387941929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-plugin-vanilla/commit/937405e6c9131647a8faa4c6b56486c387941929"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
