)]}'
{"id":"openstack%2Fsahara-tests~716213","triplet_id":"openstack%2Fsahara-tests~master~Id80ef7e186f6328ab162b92eebababd7e4886560","project":"openstack/sahara-tests","branch":"master","topic":"py2-cleanup","hashtags":[],"change_id":"Id80ef7e186f6328ab162b92eebababd7e4886560","subject":"Py2 cleanup: py3 on old branches (or none), remove py27","status":"MERGED","created":"2020-03-31 08:55:30.000000000","updated":"2020-04-16 15:12:39.000000000","submitted":"2020-04-16 15:05:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716213-1587049503653-332a2ac9","meta_rev_id":"7fc23b42f560e9ad4bd3996e818f79f067b5b25f","_number":716213,"virtual_id_number":716213,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 15:05:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-04-16 13:26:28.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-16 13:26:28.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-01 10:10:34.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-16 13:26:28.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2020-04-16 15:05:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"edfbed7be6b3e74d5f279f710047899e14c443bf","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-31 08:55:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14093144bc6c4a40594b9efd0b429a6824be29a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 10:05:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b71978ce5a949c490e07204cff1c929 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2ac2bc2d6144466811d210d78dac6f8 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a332f3c250940718c1b9710cc735688 : SUCCESS in 4m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9c1d624edf2f40f1b4e3b32b6b1a21aa : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57ba7b36a2ba424280529fa28bd8f3bd : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a49bc4e469f4a39b3eb1928f9f96ff2 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d1035c0aa05466083843f92933154e6 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0c9480e5ec64622ae1275dc92a977a1 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/440071cfc965400087ff0c768fd5c250 : SUCCESS in 3m 56s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/bb703992eb6e40f6a159494a98508c18 : FAILURE in 39m 16s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/ce3effb0da4845778f1fb88213cca035 : FAILURE in 34m 30s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/f4b6c0420280412a86b987be1985a735 : SUCCESS in 46m 17s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/0b9668ff279f4fac903bee650a9ecf7e : SUCCESS in 45m 42s\n- sahara-tests-scenario-train https://zuul.opendev.org/t/openstack/build/547f750154f3402aba2b1351f27de369 : SUCCESS in 1h 03m 36s\n- sahara-tests-scenario-stein https://zuul.opendev.org/t/openstack/build/e99b5ccf4c384ba1a9f8a46bc49d332d : FAILURE in 38m 11s\n- sahara-tests-scenario-rocky https://zuul.opendev.org/t/openstack/build/a12dc03ad0e44b9d9b8a0e2d064a4692 : FAILURE in 34m 46s\n- sahara-tests-scenario-queens https://zuul.opendev.org/t/openstack/build/dcc6ca21ca834a12b6ad0728729ac119 : FAILURE in 35m 12s\n- sahara-tests-scenario-pike https://zuul.opendev.org/t/openstack/build/5a58154758ad4b33aa848f3c37183c21 : FAILURE in 29m 47s","accounts_in_message":[],"_revision_number":1},{"id":"670f2b9689c2923b1f85e2fac03cbda036e2f256","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-31 10:24:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81cffe4b8a8560527c6f82a344a464ecf1a19031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 11:33:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f63438bff2149bf83d59ebf1b3fabaa : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/748b5701cac24c65bf80d88433daef06 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a725e5e548424cefa53f90995cf9881c : SUCCESS in 4m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/736f86456ede4fb3993005f1cd898113 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e6e4da88513451d9084ea83a4c1cacc : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e043ca0584d7411797ce65374153b078 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99d1f86ee0214580885fc32c66259726 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebc11643929e4f299fad7ffc0d81aa80 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0618e816fdfc45d69f368fe65bb35302 : SUCCESS in 3m 58s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/b5eed721560940e4acaad3219ab4e8cb : FAILURE in 34m 18s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/5a11f3c04f2e4c7babd6e368d098775f : FAILURE in 47m 00s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/4d97debc4b6e46fd9b30980a43cf4f70 : SUCCESS in 43m 41s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/3f99af97785b452bab84b80733382a40 : SUCCESS in 52m 32s\n- sahara-tests-scenario-train https://zuul.opendev.org/t/openstack/build/b8de7170b07e49e3bee907cebf15dfe3 : SUCCESS in 58m 56s\n- sahara-tests-scenario-stein https://zuul.opendev.org/t/openstack/build/c2e0c7947c4b46b080e3893cee365fa2 : SUCCESS in 1h 07m 05s\n- sahara-tests-scenario-rocky https://zuul.opendev.org/t/openstack/build/e54c23784a6344d3989557f750c9276c : FAILURE in 35m 39s\n- sahara-tests-scenario-queens https://zuul.opendev.org/t/openstack/build/ae44667deaee48c0aadab60190b2436f : FAILURE in 37m 46s\n- sahara-tests-scenario-pike https://zuul.opendev.org/t/openstack/build/747030c5f2db45dbb71693af8eb7ee62 : FAILURE in 12m 50s","accounts_in_message":[],"_revision_number":2},{"id":"7baeb9d80f3c41a05a4e014a3b82fa5f11eb85c2","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-03-31 15:23:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b59b5841692e564e3da6dcb5788316a884f85731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 16:46:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77cd09d9ee454da185c884d2d107fa7a : SUCCESS in 6m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d6c2099da1f46eb8bd0c149f1cc7289 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e131b725954f0b8b650365a57025c1 : SUCCESS in 5m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/590f901d051c4237ad3f29a2ad7d4b7b : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50fd9895464843d5aa1ceb475b242b62 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/107d7d062c3a4c71890e0b3a3c01bfc7 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/419173ec59ab4d37a45968846d28c991 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05d3a8a8238e4ec09a94e5e2472dbaa1 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8820420a836a4d64abdb1ea848b5e51f : SUCCESS in 4m 39s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/0dcafaa939714f3f877eda3d92ce0ef2 : FAILURE in 36m 10s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/140b910333264c32a35cc0442adf8900 : FAILURE in 44m 44s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/521e2cb441a94070851d81fc27d38a0c : SUCCESS in 48m 08s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/9c8890f3fcbe4a868f383cc2a2cfec5e : SUCCESS in 42m 30s\n- sahara-tests-scenario-train https://zuul.opendev.org/t/openstack/build/2854e452495640bc84e21e0c85040656 : SUCCESS in 1h 11m 12s\n- sahara-tests-scenario-stein https://zuul.opendev.org/t/openstack/build/650287a631574b2b8fb08009a33d89d1 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":3},{"id":"afc58da168775243d42c876c4bccdaf7429446ad","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-01 09:27:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6aa7cce8c8441e12cc9a6e9d6046ac8f1e02a4c4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-01 10:10:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8f27f290322b821b15d9c2de5a82c4ef34de03bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 10:30:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f644092e503048c2bea198ea17a5f646 : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61ac17a772b044ad89bcbd07582fae8a : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3950f571ab58434d840003f5e5dc81c7 : SUCCESS in 5m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/02f6d715ecd54193a7f8d1bae5295b57 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40fc951a549d405288cd34b612a4845f : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b52c116998334b5090978fedfc4c31d5 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c3b7950eca747829b6a27cdb5b44aae : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f424ef7c6a324759a955228f87b443ca : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3eab8673833f4bf197361384c1e04abb : SUCCESS in 4m 56s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/b745c4f108f640f7bad087e45fa0378d : FAILURE in 34m 53s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/cb1d02b6f7734f40a2220481f928cb2c : FAILURE in 36m 27s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/9d8bcad1aebc4b609f1b136b62ce6b26 : SUCCESS in 46m 33s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/e08654f2cdfe46c3b75be0657ffa1b4a : SUCCESS in 42m 16s\n- sahara-tests-scenario-train https://zuul.opendev.org/t/openstack/build/aaf272b4247a49acaddd9939d74fb12c : SUCCESS in 1h 00m 34s\n- sahara-tests-scenario-stein https://zuul.opendev.org/t/openstack/build/448fcafdda8b42958672629381f20fe7 : SUCCESS in 56m 32s","accounts_in_message":[],"_revision_number":4},{"id":"b034c6acabf1fc6e617b654d486fae78845e8442","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-01 12:04:30.000000000","message":"Patch Set 4:\n\nUmpf, we have https://storyboard.openstack.org/#!/story/2007501","accounts_in_message":[],"_revision_number":4},{"id":"d907c78945950e98d169cccec42b3150a4c4ec10","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-01 13:23:29.000000000","message":"Patch Set 4:\n\nI\u0027m going to send another patch, but it will still fail until https://storyboard.openstack.org/#!/story/2007501 is fixed.","accounts_in_message":[],"_revision_number":4},{"id":"d426a2b17127476e21c6985c23948f595b13a173","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-01 13:24:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f438c15a2014629ad7752b117dc4df578c1b42ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 14:39:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f641e6939e47e8ab702f6300c90b05 : SUCCESS in 4m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4ed4b4ce875c4fd5855c7180e883baac : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18388c5077dc4c1da1a9d17cad60cbb3 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fbe295b62834c67a9676d14704218d1 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab2a2cc1f4894a429366e69c816894de : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eca579b2b3a84bbcb8baf28ff5d17cb4 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b13835a03ffb47d9a7e3ba52172446f2 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/817364492c3a4b92ad53ce9527002c2b : SUCCESS in 3m 53s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/cae0a7b947e64dc5a31391c2d1e22866 : FAILURE in 42m 03s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/2415777f812541f28d31d39ae2b94015 : FAILURE in 39m 08s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/1fa1222189fe44c2a9428b96ac93083e : SUCCESS in 43m 42s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/e5b28412a8dc46cabd18cad42035f687 : SUCCESS in 41m 11s\n- sahara-tests-scenario-train https://zuul.opendev.org/t/openstack/build/2512c69a5919469aac4f9b1eb311eb2c : SUCCESS in 1h 00m 59s\n- sahara-tests-scenario-stein https://zuul.opendev.org/t/openstack/build/1f6dee44a77a474fbcd680e61c517328 : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":5},{"id":"6c2e91316c7124a66095e08e4eea2a3f0678b60e","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-10 07:37:57.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ab7f786d031eaf8459e66482195dacfa9997b51c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 08:47:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f19ea96ecc3c4a68919868c0eef139ea : SUCCESS in 5m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bcff8a2464b04ddba2d6dd9a12790afe : FAILURE in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f26b7daafb14679a198581de6ff9185 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1e37cfb58424fadb02408f7cbd1f37a : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d03a5cb8dfcb428b82238b7dc3d5ba69 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a7d592c0e2a4c65969a68b7a0cf3b68 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be9d26e685104f17bffd56ad4f8857d9 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0d111a55829347518e66919482f0cd8d : SUCCESS in 5m 55s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/a0bdd48aa5224dc99cc06847af723654 : SUCCESS in 59m 34s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/01a9c83333514b7e9f318eb5648e8538 : SUCCESS in 47m 03s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/f8259da43735406b842d0e659d82d7d6 : SUCCESS in 46m 36s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/c59221062daa4a06a1388dca406b6f3d : SUCCESS in 48m 47s\n- sahara-tests-scenario-train https://zuul.opendev.org/t/openstack/build/2ad489a71134417e8cb61312818f197a : SUCCESS in 1h 02m 07s\n- sahara-tests-scenario-stein https://zuul.opendev.org/t/openstack/build/c14e773dc85e43c985c7c1ff3d6b4df3 : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":5},{"id":"19447487d5eabc1d3a7231be4dc5d7256d63d433","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-10 08:50:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"362b061ad7a3846dd26f42706f4475351a3a810c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 09:47:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d8c43b0515413b86bff5ad5d1ab843 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa594e551c154a368474ad1489793d3d : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8026798819e4ce3a273d73f6c8ce6bb : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/451a7fa41bb142e0804f51d8b78fd5cf : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56f7a289ce8b456697a6a1af21fcf633 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d13f76bd9f0468e91550eabbec536ca : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f5bb9eb6470b454c886c9d62050b89d9 : SUCCESS in 3m 59s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/f97f34003cfc4f0d82eca09557ed020d : SUCCESS in 55m 11s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/1483b89b499342fd881b18de7e2ec159 : SUCCESS in 54m 42s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/a11d884ee68d4a5e89ffc1e2b9bd925a : SUCCESS in 43m 55s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/a5684f9362394880ba78f1fafa5602ae : SUCCESS in 40m 20s\n- sahara-tests-scenario-train https://zuul.opendev.org/t/openstack/build/22ef16ee0b3c4f67b436ad3f62461edc : SUCCESS in 55m 46s\n- sahara-tests-scenario-stein https://zuul.opendev.org/t/openstack/build/722fbc2e293b4e2cbaa64a9d12f2f3bf : SUCCESS in 54m 35s","accounts_in_message":[],"_revision_number":6},{"id":"0f78687af3e47780fe9d3bd733b95278641329ec","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2020-04-16 13:26:28.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"915a3349fb1f36cb63992eb86709d6d5025d5999","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 13:26:53.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5535b935fe3fa65a530603f56a0dbe0eef1d0066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 15:05:03.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7728752965fd4c76a8f428a8497a4f99 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ab8f94222fe451c932e4a8c55cc0777 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc74848032794ea8abb3c0ad8c9f18d2 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3533bdde440e43d3aa669d2440746f93 : SUCCESS in 4m 52s\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/263f9d9261244250ba72befbce8e5dd4 : SUCCESS in 1h 10m 43s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/ce0927069cc346b1ab13bccc2c9bfe69 : SUCCESS in 1h 27m 14s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/cc0b999509f54a2a8fc7598058cb6ff2 : SUCCESS in 1h 25m 09s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/da674e2f16c54de39ad73f6df32a60db : SUCCESS in 34m 55s","accounts_in_message":[],"_revision_number":6},{"id":"050fa7aa7352a55db0a1b29369de60772f40f329","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 15:05:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"efa3db557c5590f40564f728760cd10d5ad1b4a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 15:12:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a90d8efed9a64186a8697a0778137bf6 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2cb44abfceb48344cf3b32863d7bab01eb41312b","revisions":{"7399e5de96b2ddb0a83baa1e2d3c474a1fdbab57":{"kind":"REWORK","_number":1,"created":"2020-03-31 08:55:30.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/13/716213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-tests","ref":"refs/changes/13/716213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/1"}}},"commit":{"parents":[{"commit":"752970e295409caed918073f500dfc3926f1ce5a","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/752970e295409caed918073f500dfc3926f1ce5a"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-31 08:53:48.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-31 08:53:48.000000000","tz":120},"subject":"Python 2 cleanup: old branches should run on py3","message":"Python 2 cleanup: old branches should run on py3\n\nMake sure that the job running on master are py3 based,\naccording the established policy.\nKeep the py2 variant so that the job can be used in stable branches.\n\nChange-Id: Id80ef7e186f6328ab162b92eebababd7e4886560\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/7399e5de96b2ddb0a83baa1e2d3c474a1fdbab57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/7399e5de96b2ddb0a83baa1e2d3c474a1fdbab57"}]},"branch":"refs/heads/master"},"f7d80ef841bf696a8eb85d0a69e106953e7c8b1d":{"kind":"REWORK","_number":2,"created":"2020-03-31 10:24:09.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/13/716213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-tests","ref":"refs/changes/13/716213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/2"}}},"commit":{"parents":[{"commit":"752970e295409caed918073f500dfc3926f1ce5a","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/752970e295409caed918073f500dfc3926f1ce5a"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-31 10:24:00.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-31 10:24:00.000000000","tz":120},"subject":"Python 2 cleanup: old branches should run on py3","message":"Python 2 cleanup: old branches should run on py3\n\nMake sure that the job running on master are py3 based,\naccording the established policy.\nRGW must be used to provide the Swift API on Stein and earlier\nversions, as devstack cannot properly deploy Swift/py2\nin a Python 3 environment (old issue never fully solved).\n\nKeep the py2 variant so that the job can be used in stable\nbranches.\n\nChange-Id: Id80ef7e186f6328ab162b92eebababd7e4886560\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/f7d80ef841bf696a8eb85d0a69e106953e7c8b1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/f7d80ef841bf696a8eb85d0a69e106953e7c8b1d"}]},"branch":"refs/heads/master"},"0296a990b0305c3c88da0e3036778fee83fc0146":{"kind":"REWORK","_number":3,"created":"2020-03-31 15:23:06.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/13/716213/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-tests","ref":"refs/changes/13/716213/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/3"}}},"commit":{"parents":[{"commit":"752970e295409caed918073f500dfc3926f1ce5a","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/752970e295409caed918073f500dfc3926f1ce5a"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-31 15:19:33.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-03-31 15:19:33.000000000","tz":120},"subject":"Make sure old branches should run on py3 (or not at all)","message":"Make sure old branches should run on py3 (or not at all)\n\nMake sure that the job running on master are py3 based,\naccording the established policy.\nRGW must be used to provide the Swift API on Stein,\nas devstack cannot properly deploy Swift/py2\nin a Python 3 environment (old issue never fully solved).\n\nRemove all EM releases, where Python 3.6 is not available\nand Tempest is not supported anymore. They will be gated\nby sahara and python-saharaclient stable gates if needed;\nthe py2 variants are kept for this reason.\n\nIf an incompatible changes breaks an older branch,\nan older sahara-tests version will be pinned there.\n\nChange-Id: Id80ef7e186f6328ab162b92eebababd7e4886560\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/0296a990b0305c3c88da0e3036778fee83fc0146"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/0296a990b0305c3c88da0e3036778fee83fc0146"}]},"branch":"refs/heads/master"},"bd281586396d6dedbb317079e5bd70d964273bf1":{"kind":"REWORK","_number":4,"created":"2020-04-01 09:27:35.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/13/716213/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-tests","ref":"refs/changes/13/716213/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/4"}}},"commit":{"parents":[{"commit":"752970e295409caed918073f500dfc3926f1ce5a","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/752970e295409caed918073f500dfc3926f1ce5a"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-01 09:18:20.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-01 09:18:20.000000000","tz":120},"subject":"Make sure old branches should run on py3 (or not at all)","message":"Make sure old branches should run on py3 (or not at all)\n\nMake sure that the job running on master are py3 based,\naccording the established policy.\nRGW must be used to provide the Swift API on Stein,\nas devstack cannot properly deploy Swift/py2\nin a Python 3 environment (old issue never fully solved).\n\nRemove all EM releases, where Python 3.6 is not available\nand Tempest is not supported anymore. They will be gated\nby sahara and python-saharaclient stable gates if needed;\nthe py2 variants are kept for this reason.\n\nIf an incompatible changes breaks an older branch,\nan older sahara-tests version will be pinned there.\n\nFinally, fix a multi-line issue with a shell command\nwhich strangely happens for the master job only.\n\nChange-Id: Id80ef7e186f6328ab162b92eebababd7e4886560\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/bd281586396d6dedbb317079e5bd70d964273bf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/bd281586396d6dedbb317079e5bd70d964273bf1"}]},"branch":"refs/heads/master"},"4f6f89d8f9e0b129d6f450cec060cdc1cf8f51da":{"kind":"REWORK","_number":5,"created":"2020-04-01 13:24:17.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/13/716213/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-tests","ref":"refs/changes/13/716213/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/5"}}},"commit":{"parents":[{"commit":"752970e295409caed918073f500dfc3926f1ce5a","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/752970e295409caed918073f500dfc3926f1ce5a"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-01 13:23:47.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-01 13:23:47.000000000","tz":120},"subject":"Py2 cleanup: py3 on old branches (or none), remove py27","message":"Py2 cleanup: py3 on old branches (or none), remove py27\n\nMake sure that the job running on master are py3 based,\naccording the established policy.\nRGW must be used to provide the Swift API on Stein,\nas devstack cannot properly deploy Swift/py2\nin a Python 3 environment (old issue never fully solved).\n\nRemove all EM releases, where Python 3.6 is not available\nand Tempest is not supported anymore. They will be gated\nby sahara and python-saharaclient stable gates if needed;\nthe py2 variants are kept for this reason.\n\nIf an incompatible changes breaks an older branch,\nan older sahara-tests version will be pinned there.\n\nRemove the leftover py27 unit test jobs, which were\nstill needed when the py2 version of sahara-tests was\nused on older branches (the py3 version is always used\neven when the deployment is py2).\n\nFinally, fix a multi-line issue with a shell command\nwhich strangely happens for the master job only.\n\nChange-Id: Id80ef7e186f6328ab162b92eebababd7e4886560\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/4f6f89d8f9e0b129d6f450cec060cdc1cf8f51da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/4f6f89d8f9e0b129d6f450cec060cdc1cf8f51da"}]},"branch":"refs/heads/master"},"2cb44abfceb48344cf3b32863d7bab01eb41312b":{"kind":"REWORK","_number":6,"created":"2020-04-10 08:50:03.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/13/716213/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara-tests","ref":"refs/changes/13/716213/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara-tests refs/changes/13/716213/6"}}},"commit":{"parents":[{"commit":"752970e295409caed918073f500dfc3926f1ce5a","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/752970e295409caed918073f500dfc3926f1ce5a"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-10 08:49:17.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-10 08:49:17.000000000","tz":120},"subject":"Py2 cleanup: py3 on old branches (or none), remove py27","message":"Py2 cleanup: py3 on old branches (or none), remove py27\n\nMake sure that the job running on master are py3 based,\naccording the established policy.\nRGW must be used to provide the Swift API on Stein,\nas devstack cannot properly deploy Swift/py2\nin a Python 3 environment (old issue never fully solved).\n\nRemove all EM releases where Python 3.6 is not available\nand Tempest is not supported anymore. They will be gated\nby sahara and python-saharaclient stable gates if needed;\nthe py2 variants are kept for this reason.\nAt the same time, remove the py35 unit test job,\nas it is not supported anymore.\n\nIf an incompatible changes breaks an older branch,\nan older sahara-tests version will be pinned there.\n\nRemove the leftover py27 unit test jobs, which were\nstill needed when the py2 version of sahara-tests was\nused on older branches (the py3 version is always used\neven when the deployment is py2).\n\nFinally, fix a multi-line issue with a shell command\nwhich strangely happens for the master job only.\n\nChange-Id: Id80ef7e186f6328ab162b92eebababd7e4886560\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/2cb44abfceb48344cf3b32863d7bab01eb41312b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara-tests/commit/2cb44abfceb48344cf3b32863d7bab01eb41312b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
