)]}'
{"id":"openstack%2Fsahara~681997","triplet_id":"openstack%2Fsahara~master~I1009205ee8298065cece5f0cda228b06a1281f14","project":"openstack/sahara","branch":"master","topic":"681997","hashtags":[],"change_id":"I1009205ee8298065cece5f0cda228b06a1281f14","subject":"Fixing broken links and removing outdated driver","status":"MERGED","created":"2019-09-13 09:00:41.000000000","updated":"2019-09-25 07:21:46.000000000","submitted":"2019-09-25 07:20:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681997-1569396023849-c425b5ec","meta_rev_id":"9cbf0fd672ae6dd92aff9198027e17eaee389632","_number":681997,"virtual_id_number":681997,"owner":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"date":"2019-09-20 16:10:20.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-25 07:20:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2019-09-25 07:07:49.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":2,"date":"2019-09-20 16:12:34.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":1,"date":"2019-09-25 07:07:49.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 16:10:20.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2019-09-20 16:12:34.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-09-25 07:07:49.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-09-25 07:20:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3cef627cfd24f406ef6fddda7d48f2078e396596","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-09-13 09:00:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"047b550171986baf6c52bf6acebbc2579b738553","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 09:08:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2aae2aa7ac14158b67d5c5de31a2c45 : FAILURE in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dbc0857edf445df97602287b8cc6208 : FAILURE in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"65e4bc22cd55966146613582f45f26c372835960","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-09-13 09:18:19.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4e0cbac6c737dff25d97cbb0238adbee8ef999b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 09:26:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/097a29c9c2d14cd786ae0b82db2a08d1 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87915a074cc24ab4a6423dd72d54af07 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f8c2907100945e6c4c136b88e1c2460a736575d5","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-09-13 09:32:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c9ab034d958553d35d5e07a804528b11fecb8aa4","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-09-13 14:05:12.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3b1a624c5a8d515073448668e455dfa1e5c5f4cd","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-09-13 14:14:02.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa4b03c6760972dd7fbf2c25cf612163153e1c32","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-09-20 15:21:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cb47f90ca021ef601d745930865c21c36c34b251","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-09-20 15:29:29.000000000","message":"Patch Set 3:\n\n\u003e Uploaded patch set 3.\n\nThank you!","accounts_in_message":[],"_revision_number":3},{"id":"77110324c2d560b5a917c6e7fdbadc38a2fb5dad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 15:33:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd1e238b046d4ce8a2271e668408cce3 : FAILURE in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd61c829cbb04745afbd6852dce6a651 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":3},{"id":"84b9a8315453137a0e02a235abbfcff3b083b497","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-09-20 15:37:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ce9ab74dc297269066c8dee4774a31c3104ef414","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 15:55:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ec4b86635340eeb8fce41436ca6183 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ba10e5fa1ad409b8c138b132f565dfb : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":4},{"id":"fd51ff0c618a94eec7e0a3b0116a3facdc22e82f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-09-20 16:02:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4464b4d55ff33ab5b30c0b090e0cdcc5f6521dd4","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2019-09-20 16:10:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4e9b8842b736b0388929c1a15954c68258211255","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-09-20 16:12:34.000000000","message":"Patch Set 4: Code-Review+2\n\noh, uh, right","accounts_in_message":[],"_revision_number":4},{"id":"fdcdc41575e595e41ffcf44234baaa7e6464115a","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-09-25 07:07:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9d3136377f33e85dec3bff00aadfbd1cafb59d53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 07:08:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d7eff6d67533085f8f70afeaa15c7c5c0ad41465","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 07:20:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f432518c08bb4165813b5bf84e1a8bd0 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5357385f5254a9680307e96a8ebde3a : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":4},{"id":"81e79dbb5f93614a5985318c0b214e96572dedcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 07:20:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9e29c64f8170821b9ed0782bcf888d1847af184d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 07:21:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a91777b48d4e859264db2affa801be : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6e3e4b91c2068f7db3b360286560b1d19636856f","revisions":{"bb3cc0844cf2cb4dcacf4e36d874e8431047f568":{"kind":"REWORK","_number":1,"created":"2019-09-13 09:00:41.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/97/681997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/97/681997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/97/681997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/97/681997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/97/681997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/97/681997/1"}}},"commit":{"parents":[{"commit":"7e3184c561a34a4e586a6dc973450c1595841b18","subject":"Merge \"Remove a monkey-patching workaround for python \u003c 2.7.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/7e3184c561a34a4e586a6dc973450c1595841b18"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-09-13 08:59:50.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-09-13 08:59:50.000000000","tz":60},"subject":"Fixing broken links and removing outdated driver","message":"Fixing broken links and removing outdated driver\n\nzmq is deprecated, removing unnecessary content\n\nChange-Id: I1009205ee8298065cece5f0cda228b06a1281f14\nBackport: stein rocky\nSigned-off-by: Alexandra Settle \u003casettle@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/bb3cc0844cf2cb4dcacf4e36d874e8431047f568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/bb3cc0844cf2cb4dcacf4e36d874e8431047f568"}]},"branch":"refs/heads/master"},"3fe406314036d2082f52d9f57c4d7e50568a1b94":{"kind":"REWORK","_number":2,"created":"2019-09-13 09:18:19.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/97/681997/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/97/681997/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/97/681997/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/97/681997/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/97/681997/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/97/681997/2"}}},"commit":{"parents":[{"commit":"7e3184c561a34a4e586a6dc973450c1595841b18","subject":"Merge \"Remove a monkey-patching workaround for python \u003c 2.7.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/7e3184c561a34a4e586a6dc973450c1595841b18"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-09-13 08:59:50.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-09-13 09:18:17.000000000","tz":0},"subject":"Fixing broken links and removing outdated driver","message":"Fixing broken links and removing outdated driver\n\nzmq is deprecated, removing unnecessary content\n\nChange-Id: I1009205ee8298065cece5f0cda228b06a1281f14\nBackport: stein rocky\nSigned-off-by: Alexandra Settle \u003casettle@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/3fe406314036d2082f52d9f57c4d7e50568a1b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/3fe406314036d2082f52d9f57c4d7e50568a1b94"}]},"branch":"refs/heads/master"},"3d838110c1e550f0a30db24efa8ed442b50bc778":{"kind":"REWORK","_number":3,"created":"2019-09-20 15:21:59.000000000","uploader":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"ref":"refs/changes/97/681997/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/97/681997/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/97/681997/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/97/681997/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/97/681997/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/97/681997/3"}}},"commit":{"parents":[{"commit":"7e3184c561a34a4e586a6dc973450c1595841b18","subject":"Merge \"Remove a monkey-patching workaround for python \u003c 2.7.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/7e3184c561a34a4e586a6dc973450c1595841b18"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-09-13 08:59:50.000000000","tz":60},"committer":{"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","date":"2019-09-20 15:21:58.000000000","tz":-240},"subject":"Fixing broken links and removing outdated driver","message":"Fixing broken links and removing outdated driver\n\nzmq is deprecated, removing unnecessary content\n\nChange-Id: I1009205ee8298065cece5f0cda228b06a1281f14\nBackport: stein rocky\nSigned-off-by: Alexandra Settle \u003casettle@suse.com\u003e\nCo-Authored-By: Jeremy Freudberg \u003cjeremyfreudberg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/3d838110c1e550f0a30db24efa8ed442b50bc778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/3d838110c1e550f0a30db24efa8ed442b50bc778"}]},"branch":"refs/heads/master"},"6e3e4b91c2068f7db3b360286560b1d19636856f":{"kind":"REWORK","_number":4,"created":"2019-09-20 15:37:21.000000000","uploader":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"ref":"refs/changes/97/681997/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/97/681997/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/97/681997/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/97/681997/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/97/681997/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/97/681997/4"}}},"commit":{"parents":[{"commit":"7e3184c561a34a4e586a6dc973450c1595841b18","subject":"Merge \"Remove a monkey-patching workaround for python \u003c 2.7.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/7e3184c561a34a4e586a6dc973450c1595841b18"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-09-13 08:59:50.000000000","tz":60},"committer":{"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","date":"2019-09-20 15:37:18.000000000","tz":-240},"subject":"Fixing broken links and removing outdated driver","message":"Fixing broken links and removing outdated driver\n\nzmq is deprecated, removing unnecessary content\n\nChange-Id: I1009205ee8298065cece5f0cda228b06a1281f14\nBackport: stein rocky\nSigned-off-by: Alexandra Settle \u003casettle@suse.com\u003e\nCo-Authored-By: Jeremy Freudberg \u003cjeremyfreudberg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/6e3e4b91c2068f7db3b360286560b1d19636856f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/6e3e4b91c2068f7db3b360286560b1d19636856f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
