)]}'
{"id":"openstack%2Fsahara~695913","triplet_id":"openstack%2Fsahara~stable%2Fqueens~If8ff04b6a59f98700f1aa3988fe740959a0ec4b0","project":"openstack/sahara","branch":"stable/queens","hashtags":[],"change_id":"If8ff04b6a59f98700f1aa3988fe740959a0ec4b0","subject":"Run sahara-scenario using Python 3","status":"MERGED","created":"2019-11-25 14:13:16.000000000","updated":"2019-11-26 16:51:22.000000000","submitted":"2019-11-26 16:51:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695913-1574787082707-81677534","meta_rev_id":"152979e2ca1cdd21872a049c0998530f733309b2","_number":695913,"virtual_id_number":695913,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"date":"2019-11-25 17:39:39.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 16:51:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2019-11-26 15:37:39.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":1,"date":"2019-11-26 15:37:39.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 15:37:39.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2019-11-26 16:51:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1208e414098e0fe4ecb067b0c2d0b0edde518a8f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-25 14:13:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2a83befb43941e6f159d7bf955e4f9fecbc2089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 15:36:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdccd1feaf064f0fb86226d14656b13c : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3f43d8005f3410b90056efc4b1e8934 : SUCCESS in 7m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ffb69aa7e4534301935140b5e47692b8 : SUCCESS in 6m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b95d50ffdc2044a989621645d43ca9eb : SUCCESS in 6m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cdcf5b631b22401f83316f043faeedb8 : SUCCESS in 6m 39s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/c777013d7fa74b2eb714f676633091f3 : SUCCESS in 53m 49s\n- sahara-tests-scenario-modwsgi https://zuul.opendev.org/t/openstack/build/940abda1562245b9996b6746ca5144a8 : SUCCESS in 1h 03m 48s (non-voting)\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/0d2a124aa5cd46fa8e501db1d4743e3b : SUCCESS in 54m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f2619a2779b453fa11f0b33d1898789 : SUCCESS in 7m 57s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/5008afb28e69495ca4dee6ed08faacb1 : SUCCESS in 1h 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0e86f1960a4e8f0fc1c0f9534b4e378e5ada530b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-25 15:45:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b1d138fe2997895b20547e9332f7e71a0b3d3be7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 16:58:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c936e6c52d84714b69bdd4f7cedd98b : SUCCESS in 6m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0071589c313949a3b18918669b9c82e3 : SUCCESS in 8m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cd80c61e17ef44ba9ce98c10882e6622 : SUCCESS in 7m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a09e88b932d74934990b482a7d665982 : SUCCESS in 6m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/27c2218e19624117a6f2b2d1321cf8f9 : SUCCESS in 6m 18s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/fe22efc9c3a64351bd87e0214e0e55b3 : SUCCESS in 51m 49s\n- sahara-tests-scenario-modwsgi https://zuul.opendev.org/t/openstack/build/5a6eec48feb84760a71abcf3b3f1a06f : SUCCESS in 1h 08m 42s (non-voting)\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/333095b60df54f75b5c154a491175044 : SUCCESS in 50m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/051bd2a0d7cf49c68bb6ef566950ca51 : SUCCESS in 8m 06s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/ee38e19602ce4671a67e2959c5b54ac6 : FAILURE in 44m 56s","accounts_in_message":[],"_revision_number":2},{"id":"40c175e950eb4a6a2fdcce201af945991d8903d6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-25 17:39:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2f427661505e0e736df705ea7c8bcd2bbc564b01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 18:50:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b16a11b5db24b2298ee4bffbf46d5a7 : SUCCESS in 7m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5ff83bfac3d4852b57e31007fd832f9 : SUCCESS in 7m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cadf185c7c8c484dad38c33566a6d5c9 : SUCCESS in 6m 48s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5b5041b07951470fb22df88fa3c9c35d : SUCCESS in 6m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/29d1fcc8a9d64650b60c9c6efd706afc : SUCCESS in 6m 18s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/bd1ef9a7f00242dba1a8408a172ad5a3 : SUCCESS in 52m 29s\n- sahara-tests-scenario-modwsgi https://zuul.opendev.org/t/openstack/build/f79840c171b1450a9d4026e9b6044d0b : SUCCESS in 49m 24s (non-voting)\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/836498ac64a84b6f832aacd620e3ac50 : SUCCESS in 40m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f784f0c9f5744461b7b3b6cab747515c : SUCCESS in 10m 41s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/50a36e54749d4466a7ef125e41ac8b7d : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":2},{"id":"b428aa23a2a4744adff1d9bd614d5867c74545be","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2019-11-26 15:37:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5a09129b840fc4f9bb16747eb338297ba746dfd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:38:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5f5b81296624877f36feb6e7240fcfe071df7dda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:51:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"152979e2ca1cdd21872a049c0998530f733309b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:51:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57144a2c188149de834373a823f2e0ba : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71227c0c826d48bd92e090681fc7c190 : SUCCESS in 5m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/250da38f63c24c6db8a3fe7fd2d3876d : SUCCESS in 6m 21s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/659fecf9254042328d64f9ad0eba8d35 : SUCCESS in 6m 12s\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/c124c1869d7d49b4b4ef77991e82c0ce : SUCCESS in 57m 10s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/5368e460de6245dc9da2eddea31d20d3 : SUCCESS in 54m 43s\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/22e570147a64401fb680db94515577a5 : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec5de3aec755e1b4578d4dc7659036b3324646f8","revisions":{"360ef5cbc23dd02ebea9fa4476c2787d6d5e452a":{"kind":"REWORK","_number":1,"created":"2019-11-25 14:13:16.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/13/695913/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/13/695913/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/13/695913/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/13/695913/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/13/695913/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/13/695913/1"}}},"commit":{"parents":[{"commit":"f6a07610a89fb61c41e2638fc1edc8e27be08f0d","subject":"Fixing image creation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/f6a07610a89fb61c41e2638fc1edc8e27be08f0d"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-11-21 16:48:23.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-11-25 14:09:04.000000000","tz":60},"subject":"Run sahara-scenario using Python 3","message":"Run sahara-scenario using Python 3\n\nThe sahara-tests repository is removing Python 2 support in master,\nAs Python 3.5 is supported, sahara-tests-scenario jobs can safely\nuse it to run sahara-scenario also on this stable branch.\n\nChange-Id: If8ff04b6a59f98700f1aa3988fe740959a0ec4b0\n(cherry picked from commit 8677168b80aeee669d468a09bfb754498f81f5c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/360ef5cbc23dd02ebea9fa4476c2787d6d5e452a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/360ef5cbc23dd02ebea9fa4476c2787d6d5e452a"}]},"branch":"refs/heads/stable/queens"},"ec5de3aec755e1b4578d4dc7659036b3324646f8":{"kind":"REWORK","_number":2,"created":"2019-11-25 15:45:23.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/13/695913/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/13/695913/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/13/695913/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/13/695913/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/13/695913/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/13/695913/2"}}},"commit":{"parents":[{"commit":"f6a07610a89fb61c41e2638fc1edc8e27be08f0d","subject":"Fixing image creation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/f6a07610a89fb61c41e2638fc1edc8e27be08f0d"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-11-25 15:44:09.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-11-25 15:44:09.000000000","tz":60},"subject":"Run sahara-scenario using Python 3","message":"Run sahara-scenario using Python 3\n\nThe sahara-tests repository is removing Python 2 support in master,\nAs Python 3.5 is supported, sahara-tests-scenario jobs can safely\nuse it to run sahara-scenario also on this stable branch.\n\nChange-Id: If8ff04b6a59f98700f1aa3988fe740959a0ec4b0\n(cherry picked from commit 8677168b80aeee669d468a09bfb754498f81f5c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/ec5de3aec755e1b4578d4dc7659036b3324646f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/ec5de3aec755e1b4578d4dc7659036b3324646f8"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
