)]}'
{"id":"openstack%2Fsahara~721613","triplet_id":"openstack%2Fsahara~master~If0b47fa815ef778d58542f4c7d84218aef833802","project":"openstack/sahara","branch":"master","topic":"sqlalchemy-skip","hashtags":[],"change_id":"If0b47fa815ef778d58542f4c7d84218aef833802","subject":"(Temporarily) skip TestVerifications tests","status":"MERGED","created":"2020-04-21 13:24:36.000000000","updated":"2020-04-22 16:35:22.000000000","submitted":"2020-04-22 16:27:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721613-1587572827938-81340386","meta_rev_id":"b0dc2db3d538d580c6cf1e00633fc982942690e8","_number":721613,"virtual_id_number":721613,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 16:27:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":2,"date":"2020-04-22 15:06:27.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":2,"date":"2020-04-21 19:11:56.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"all":[{"value":1,"date":"2020-04-22 15:06:27.000000000","_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-21 19:11:56.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"},{"updated":"2020-04-22 15:06:27.000000000","updated_by":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"reviewer":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"state":"REVIEWER"},{"updated":"2020-04-22 16:27:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e61a84d40d75eff4643f1e8ef01c01c0d51e9196","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-21 13:24:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c3a04ca7a5ef0ec741244862d346bbbae546a65","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-04-21 13:25:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a58a3973dfd70c2fcf96c206dbfd0fd7c55312b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 14:37:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a63886c12bb4dae9373170a5f6b2a29 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a164f1f01d54cd09a8067401598cbb4 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/892fbf5fab494732974b4d5ce0d5628a : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2770afb4d1e471c8b7cbaee1b227b58 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0419bc451044864a3a0e27c0d8a2ea3 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2916bfbde604f2597a938e144f3b3b3 : SUCCESS in 6m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a8beb879a55845f596e7862f0e5e563f : SUCCESS in 6m 40s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/fd7dc37be3914b889b6c9ee3491d1dfc : SUCCESS in 1h 04m 15s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/471b25517e5149468910f2648fc234c1 : SUCCESS in 54m 34s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/43a2d739b6954ba0853b1025224d0441 : SUCCESS in 46m 52s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/ef8b1292db4444109053dff2b6480c1a : SUCCESS in 49m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7eb5483ee3be4981b5e57d52f871f5ac : FAILURE in 7m 50s (non-voting)\n- sahara-tests-scenario-py3 https://zuul.opendev.org/t/openstack/build/8d843997fa2f48fc826c696c52098d60 : SUCCESS in 57m 40s (non-voting)\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/7d02e4483bfd448bbe745a4b81f2b32b : SUCCESS in 41m 08s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/settings","accounts_in_message":[],"_revision_number":2},{"id":"53f30bf8e023b2887b2ebc53704191cfd29f7f2f","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2020-04-21 19:11:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9d48f4f96877c1463373e055163f2e89414f40ae","author":{"_account_id":8932,"name":"Telles Mota Vidal Nóbrega","email":"tellesnobrega@gmail.com","username":"tellesnobrega"},"date":"2020-04-22 15:06:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"238f1ec7ae2076f7995d6566bd6e97c2eeac70ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 15:06:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0b35140059a24db464b4104787f35348b7c21683","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 16:27:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b8f3418ddf5935eac7636b24520a1453b1751072","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 16:27:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a558bc0d2c374d55bd6e06f642640134 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9631415ecc83421993a52a41d8e24318 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/315a731c893f4fc6ab5ffa108f0cec6f : SUCCESS in 9m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b1b01cd9ab045d3a56222dbd8ac3fad : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e2ed394a1b480bb00e6d4d658070d8 : SUCCESS in 6m 19s\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/24bcdec8e84c4848898019fa3aaefea2 : SUCCESS in 1h 11m 15s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/123c0ff158d4466ba40a34dabb635e6a : SUCCESS in 1h 06m 16s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/f36e1aa9c012432393169c49046dc884 : SUCCESS in 56m 05s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/71aff277b6ed479fa7cf3cc9ad66e89e : SUCCESS in 44m 32s\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/db4ded9c420043aa9ffe6baa0870c76d : SUCCESS in 44m 10s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/settings","accounts_in_message":[],"_revision_number":2},{"id":"d57353b0808d80c672fea506c28f23077d8ad36d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 16:35:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13fec9ea767f4672a0d0ad1124f1a0d0 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c33675508ce7e5b9b3170eccceab4569347fe04f","revisions":{"7bcd92c5c7f3745d50dbb045d1a1d8e9d36b9257":{"kind":"REWORK","_number":1,"created":"2020-04-21 13:24:36.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/13/721613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/13/721613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/13/721613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/13/721613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/13/721613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/13/721613/1"}}},"commit":{"parents":[{"commit":"0825bddef65c11928ae6d37a2a7702e2d9435143","subject":"Fix syntax error in image widths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/0825bddef65c11928ae6d37a2a7702e2d9435143"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-21 13:23:01.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-21 13:23:01.000000000","tz":120},"subject":"(Temporarily) skip TestVerification tests","message":"(Temporarily) skip TestVerification tests\n\nThey are failing thanks to sqlalchemy.orm.exc.DetachedInstanceError:\nParent instance \u003cClusterVerification at ...\u003e is not bound to a Session;\nlazy load operation of attribute \u0027checks\u0027 cannot proceed\n(Background on this error at: http://sqlalche.me/e/bhk3)\n\nThe issue seems to be difficult to tackle, so let\u0027s reevaluate after\ncutting Ussuri branches.\n\nStory: 2007450\nTask: 39508\nChange-Id: If0b47fa815ef778d58542f4c7d84218aef833802\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/7bcd92c5c7f3745d50dbb045d1a1d8e9d36b9257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/7bcd92c5c7f3745d50dbb045d1a1d8e9d36b9257"}]},"branch":"refs/heads/master"},"c33675508ce7e5b9b3170eccceab4569347fe04f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-21 13:25:30.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/13/721613/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/13/721613/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/13/721613/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/13/721613/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/13/721613/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/13/721613/2"}}},"commit":{"parents":[{"commit":"0825bddef65c11928ae6d37a2a7702e2d9435143","subject":"Fix syntax error in image widths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/0825bddef65c11928ae6d37a2a7702e2d9435143"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-21 13:25:19.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-04-21 13:25:19.000000000","tz":120},"subject":"(Temporarily) skip TestVerifications tests","message":"(Temporarily) skip TestVerifications tests\n\nThey are failing thanks to sqlalchemy.orm.exc.DetachedInstanceError:\nParent instance \u003cClusterVerification at ...\u003e is not bound to a Session;\nlazy load operation of attribute \u0027checks\u0027 cannot proceed\n(Background on this error at: http://sqlalche.me/e/bhk3)\n\nThe issue seems to be difficult to tackle, so let\u0027s reevaluate after\ncutting Ussuri branches.\n\nStory: 2007450\nTask: 39508\nChange-Id: If0b47fa815ef778d58542f4c7d84218aef833802\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/c33675508ce7e5b9b3170eccceab4569347fe04f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/c33675508ce7e5b9b3170eccceab4569347fe04f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
