)]}'
{"id":"openstack%2Fsahara~725077","triplet_id":"openstack%2Fsahara~master~Id398eb0310b53a7004f6e0596a3135dc84c1a06b","project":"openstack/sahara","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Id398eb0310b53a7004f6e0596a3135dc84c1a06b","subject":"Imported Translations from Zanata","status":"MERGED","created":"2020-05-03 08:31:18.000000000","updated":"2020-05-03 10:11:27.000000000","submitted":"2020-05-03 10:10:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725077-1588500608062-b9db8533","meta_rev_id":"c5bd74dd835355510f8f08300b721b937ebc8937","_number":725077,"virtual_id_number":725077,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-03 10:10:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":2,"date":"2020-05-03 09:59:27.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":1,"date":"2020-05-03 09:59:27.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-03 09:59:27.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-05-03 10:10:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c67bc7df9a37da4c7cda033ed3717418027a91fc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-03 08:31:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5c841abebdea32641436e43940826b867b0c525","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 08:45:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d64f9acc371410c97ae24390165ddb9 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab7012ec8194a29b1c9de9ec8679e49 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2a92d8676734bb788ae83b3360eec7c : SUCCESS in 13m 38s\n\nWarning:\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/plugin.sh","accounts_in_message":[],"_revision_number":1},{"id":"3f00c48263ce1fd4f4487c2bc9df2b9f22c94715","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-03 09:59:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3df2260f9afe539559527bf1bf858258de6dfbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 09:59:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fef9df658de333c958ca2b03e5517970f6822ac9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 10:10:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e3d63c56b145e3b9c8689d8a7883f0 : SUCCESS in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921b4a9a76564925a0c23b5df31a8fa7 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e10eda5582e47f08ac041bfe3205597 : SUCCESS in 8m 47s\n\nWarning:\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/plugin.sh","accounts_in_message":[],"_revision_number":1},{"id":"c0f897f9c15a9e4862d4542f52fd62f72d5e5dab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 10:10:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9075e7f2057e96cf18e82e3233f87b29d8a76830","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 10:11:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbfb4f73e7d4ff48bb9731b6a0bf6ed : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1cfc8f8fa7b49a1a866b05ce2d45847 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f07476d7572d1ffc76d03b7588c7a5e5f90ff421","revisions":{"f07476d7572d1ffc76d03b7588c7a5e5f90ff421":{"kind":"REWORK","_number":1,"created":"2020-05-03 08:31:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/725077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/77/725077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/77/725077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/77/725077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/77/725077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/77/725077/1"}}},"commit":{"parents":[{"commit":"89ad19ea16c3f37c2b6e1e2b7bb653458b902fe0","subject":"Merge \"Bump default tox env from py37 to py38\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/89ad19ea16c3f37c2b6e1e2b7bb653458b902fe0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-03 08:31:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-03 08:31:16.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Id398eb0310b53a7004f6e0596a3135dc84c1a06b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/f07476d7572d1ffc76d03b7588c7a5e5f90ff421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/f07476d7572d1ffc76d03b7588c7a5e5f90ff421"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
