)]}'
{"id":"openstack%2Fsahara~765482","triplet_id":"openstack%2Fsahara~master~I0be28c20ff21805d9a1a52b702cf497df7b9069e","project":"openstack/sahara","branch":"master","topic":"invalid_message","hashtags":[],"change_id":"I0be28c20ff21805d9a1a52b702cf497df7b9069e","subject":"Fix the isolated exception message","status":"MERGED","created":"2020-12-04 07:21:30.000000000","updated":"2021-01-28 12:58:01.000000000","submitted":"2021-01-28 12:56:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765482-invalid_message","meta_rev_id":"13781cb63be23b701a4baeadc02b81774b7411e9","_number":765482,"virtual_id_number":765482,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-28 12:56:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"all":[{"value":2,"date":"2020-12-14 18:13:44.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":2,"date":"2020-12-05 10:39:44.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"all":[{"value":1,"date":"2020-12-14 18:13:44.000000000","_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 08:56:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 17:17:01.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2020-12-05 10:39:44.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-12-14 18:13:44.000000000","updated_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"reviewer":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"state":"REVIEWER"}],"messages":[{"id":"b90c3e34ceddc2bfc055d0fda4b43bbe14bd341c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-12-04 07:21:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41c95642d82c9069cd3ceb52f5f11d2c806434b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 08:56:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc39adbda1054456baef24ca1c4cb0d9 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c3cecfb4864454e8286a83a944190e3 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85395f846b2d4ebdb99b01f414fd9d24 : SUCCESS in 9m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b053f5ec74a341458f2f48d6e0e2544f : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d98bf88d9fd94be8a5c162470a6bcd8b : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa962573348428ea76feeb6ba8ff312 : SUCCESS in 6m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7ff0d2b987654571b25faaf1a1b40930 : FAILURE in 6m 10s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/2842b9cd3254480eb9d977cc8b12db68 : FAILURE in 35m 36s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/1dc8917c9b464143876c97a7e9ddb454 : FAILURE in 41m 02s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/3ba607def2dc4d0e97daf35781e55e7e : FAILURE in 42m 48s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/b6c5d1c3c964452ebbc5b502f2a45346 : FAILURE in 32m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87a2281cf1ca449b886d0f59340495d4 : SUCCESS in 8m 22s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/3173940a8d944697b15f2c00431fe07a : FAILURE in 49m 18s\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/a84d16243f8249978c31a54410650866 : SUCCESS in 45m 06s\n\nWarning:\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings","accounts_in_message":[],"_revision_number":1},{"id":"a94250f22f0ea77b69b1e686911585d6a9c63ac4","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-12-04 09:12:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5b320715631d0e6fed581ade7001e971c2032df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 10:16:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02f6b4569a2b4ff9bdc8b39de2e4b773 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da687b4375904c0ea806fdc47cc0bd6b : SUCCESS in 10m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b722e5946ace4750b17f91c284a4feca : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf500c8d1ff049d18ea16d2b271dc12b : SUCCESS in 11m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/283415a56d8c4207a9b9aa5b4a32891d : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f14e68e68a4cfd884433debe2a0074 : SUCCESS in 9m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83ee9a24f6874ad7b17d0e50326bb6ed : FAILURE in 5m 59s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/da2e24dd90354dfba01f51ad088449d3 : FAILURE in 33m 36s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/8f6fe51dc5f24cdc94e65e85c519a157 : FAILURE in 49m 32s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/088ddc7334f142739a4446b093cf7a1f : FAILURE in 41m 35s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/65d05681827342db801d83cfc1ffe116 : FAILURE in 37m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43c011d11b624b23b5d8422f1b871ad6 : SUCCESS in 6m 42s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/4051ff2e6b0b4085aa89fb53b501c0b5 : FAILURE in 45m 23s\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/954e5350c6904b87a1bcf57f4f6acaf0 : SUCCESS in 58m 16s\n\nWarning:\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings","accounts_in_message":[],"_revision_number":1},{"id":"7e2e7924c5290714ce51828f0aa382d2988bf428","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-12-04 17:17:01.000000000","message":"Patch Set 1:\n\nrecheck\n\ndevstack failure fixed, heat issue temporarily workarounded","accounts_in_message":[],"_revision_number":1},{"id":"cbba0e8ce26d6ba4d349f89c47b4c0a07e6e6d14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 19:32:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1dfb5d8d52b4faaa775ae2882bdaa5e : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d969cca3aa42b6b14ad47a64dca048 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9037eb215627448d8d4b35368aed2654 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c4a094971e9440991bc14cb6481eda6 : SUCCESS in 9m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/638d8f686ece453fba14e720811b9ab5 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3711c1337564fa4b50b122d0e2b001b : SUCCESS in 6m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4cc42fdae7e24f65b74941545f543ffa : FAILURE in 16m 52s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/749b013aedea4de5955b01b9c7d83767 : TIMED_OUT in 2h 02m 54s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/436113ade1784a68b3d467f81c49b482 : SUCCESS in 49m 23s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/73af02e8ed3c4acf8edcba18ec8c2a0a : SUCCESS in 47m 09s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/af62361e2dca4e74abfb141e747d5c93 : SUCCESS in 37m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a24f659b2e4f4ed3b4f397f1d44dd13a : SUCCESS in 7m 19s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/3ca2daa29c6f48e2a8cb439841b39187 : FAILURE in 54m 05s (non-voting)\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/2f81a9a43cd84dc08c44b0b78297bbea : SUCCESS in 55m 18s\n\nWarning:\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings","accounts_in_message":[],"_revision_number":1},{"id":"bb117d245e6e1c99fecb909e0785dbb0d50e6e6c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-12-04 20:35:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"984fe40cdcf06940a6f0c6ae734aa15f335a73da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 22:04:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddbc34dc6bc04d738ff5f7beddb3bfc9 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e769064b29474b9c80f39e05b158f804 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02c1fdc7876f4e0ab2a1a9439a1881fc : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21d8ad7d6ea0417f8ccaa0446981ada4 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61615a8cc7de4218a9d65046f5431936 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/719ad5c558e34739b9eba5095dbd91f0 : SUCCESS in 5m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3e13a3756c3d4d85a223ecfd99c23b11 : FAILURE in 5m 12s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/cc3421e6cddd4dcd89239743ee4dfa7d : SUCCESS in 49m 09s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/761444ee771d432ca35c2d75100035f9 : SUCCESS in 49m 11s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/eb7b9649489847428f0b43a2fc495bfe : SUCCESS in 41m 13s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/f3bc076ecaa14069bd057da7345a495d : SUCCESS in 43m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c75ae37be3a14be396349645f9042d17 : SUCCESS in 6m 52s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/098da0278f604677b703410354540a09 : FAILURE in 41m 45s (non-voting)\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/aa5f68dffd4048e9b02033847b5822e7 : SUCCESS in 47m 03s\n\nWarning:\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings","accounts_in_message":[],"_revision_number":1},{"id":"4765a0b54661bae3a7bd366a52151aaf661ed590","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-12-05 10:39:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8fac0429f2aaa68e5aabd2bedfd34b60bf07bfb6","author":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"},"date":"2020-12-14 18:13:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2fa5f443b4c7afe7d5e19e09d13bc064751f8681","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 18:13:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cc956ab116de3e74f9e1cc099a25f79587d3aa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 19:08:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2910093572444121996695ffba73345c : FAILURE in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a126537317441b876f229409bc2d92 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0389e79f095546b098b74902ee9a5d84 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8cfaf171cc1452ab044b8c86d9d4806 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce31ba85f247457e9619a1b9bef2c5be : SUCCESS in 7m 00s\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/4359890ad5b546efb9aa47d23dc9f755 : SUCCESS in 37m 43s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/4e7000604f944db7af98b5bc75aa55f9 : SUCCESS in 42m 28s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/4f2ff5431cfa403e9c2256caaf03d390 : SUCCESS in 48m 00s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/dd04c681e7dd4b0ab89e8caabaa1c5d0 : SUCCESS in 47m 58s\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/25b91902667d4c29b65b3997d1ad73ed : SUCCESS in 48m 33s\n\nWarning:\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings","accounts_in_message":[],"_revision_number":1},{"id":"ebf36719201d1a823f96b8542016625be8ecd636","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-12-15 00:58:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"172c69ed9e553f33213902ed5ab74d03ff776cbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 02:49:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ed1c4e0f8124176b619569da1ef51c5 : FAILURE in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68d7076de5904e018b9f3942d8148675 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c23b7f7653b4aab8402727a50b4b7fe : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e0848d05faf4ef99876740c8cfb4614 : SUCCESS in 8m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d2bb40f4f1649e2b5e8dab7cdc9dccd : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc5619de1bb4cd2a25edc32276f9804 : SUCCESS in 5m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/72c10320a53c40dcbdb59623622892fd : FAILURE in 9m 39s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/c846346b5a36449fba1af5295ce22b6e : SUCCESS in 1h 06m 51s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/c0530ed056944a24b6e53820d5007951 : SUCCESS in 56m 55s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/a44706bb8354430a8102191a35128df6 : SUCCESS in 48m 27s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/b6a275a004d24ed5b3cd15dc281de4f5 : SUCCESS in 44m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66b4856d978846ad943a5ad8c432c6ad : SUCCESS in 7m 51s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/e3a537938f4e49e7bd0e77a5844c6dc0 : SUCCESS in 1h 45m 29s (non-voting)\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/cccd594a355b4254ac2d5e64e2979acc : SUCCESS in 51m 13s\n\nWarning:\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings","accounts_in_message":[],"_revision_number":1},{"id":"3a9f33d3d40d8d0a1ed7b1c647712d162c782bf8","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-12-15 08:38:57.000000000","message":"Patch Set 1:\n\nuhm, the lowet constraint job needs a fix in a separate.review (the new pip resolver highlighter several issues on various projects)","accounts_in_message":[],"_revision_number":1},{"id":"7b80af568db823d2e60d817f79a562fd36e39982","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-01-28 00:06:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3cbe70cc0dce08bc02fb79b0d259a57a36e82637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 01:38:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d05cd7cc2ef4b428e6d7be5557ea957 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da7d6e7ebc544a27a044f1fc46bcfc27 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/671b87d732654e889aff161af51fa413 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bc497713f80473f9105ef923d01a104 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d6a0217f515448d958d53bda2f38837 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/880491d25430447f95b16790aeec7cdc : SUCCESS in 6m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/694a174093e24804ab04bb94b8fd2937 : FAILURE in 6m 31s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/748b7284ad03466eb69f4fd6cd772e52 : SUCCESS in 52m 01s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/cf3e48d99a69468bae45a995a004326d : SUCCESS in 1h 01m 32s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/21958fc0939c4252aec1ff382baa01a7 : SUCCESS in 40m 39s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/c757418c328649538597f716ba02096c : SUCCESS in 35m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aac69f90a23246069fc62c858c46a90a : SUCCESS in 9m 04s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/4b73e501db3f4597ac71742ff13f72ed : SUCCESS in 1h 29m 54s (non-voting)\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/3a418422f3ed4d1d8ec45fb13276df87 : SUCCESS in 46m 13s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":1},{"id":"41f8cf22c9db2b9f0abc26fb0d28051257d11016","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 01:38:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66d1fd6326d360403026caefcc441dafa547113b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 03:42:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96d4864b48c24ca490dc9cb694a368b8 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56928d30a19745dd8535aa5b75f2955e : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07394c83a3c84473b7b17ec6e12c2730 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a193bc10e53b4c6ea40b2a3b55a8d6bd : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792b43a342dd4b748bb4199fc1fc7c2d : SUCCESS in 6m 54s\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/f084ea6732d04d1e9e3dac90108d4297 : SUCCESS in 55m 03s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/0b483b85cc5c473aab78da8b3d22a538 : TIMED_OUT in 2h 02m 31s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/16a510ad03b64fd1b5595c751bcd0cc3 : SUCCESS in 39m 56s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/56c97f6492c34b14ac67ea828d00c156 : SUCCESS in 35m 57s\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/9c720b0b167243c2be7d986ed77c80f9 : SUCCESS in 48m 16s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":1},{"id":"55032b8b151b3b837355f4ada87fe80db8365f70","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-01-28 08:36:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5019cda443200d48a990d52e169b27ae8fa28780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:17:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e3c5fd5cbe542a196ad1f126b35adc3 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ccf2843c184436b92743512adb17b43 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed35408cbbaa44d38ed3ccf4d0012297 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64f968daba0c4e498953439e47acbfa7 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2e63f7533c04167b05ca9a7e55d1d2f : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41fb28dc073a44cda2a8963131b1dd67 : FAILURE in 5m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/215aa9afd7fe400b97443f0c403982d3 : FAILURE in 9m 57s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/34527df70eaf4bdfb61fc0025479ec90 : SUCCESS in 1h 11m 52s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/8482ff67784e4eceacc4b52bbcd8b866 : SUCCESS in 58m 36s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/b562e2a503c249229ddf9c987c7a2c12 : SUCCESS in 50m 53s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/d0699b1ae0e2477b93927ae5ac2a6db5 : SUCCESS in 47m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0d4432c012f4ae18f34baa8a88f1958 : SUCCESS in 8m 48s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/aeaa90de03284d96988904323bd11ac3 : SUCCESS in 1h 17m 14s (non-voting)\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/0dac419db1804edea80157a0576b7053 : SUCCESS in 45m 13s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":1},{"id":"0d9e1d5f549cf803d2647326190bd764a703ffde","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-01-28 10:22:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d4e182c285f10f52d5b599837f20acccf8190bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 11:56:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0389699a85034b28bcb65e5b17aa4526 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc99d8fbea2345f68e65fa2694b2774a : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4971297b36c94cb7a5d73bf0bd495199 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1816b9f9a6844ea9baef2254e326b706 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/305a91ea40dc4533bebcb60661fea165 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f56566fb3be460f98c8bdfa0d3d2577 : SUCCESS in 5m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/863283087d1a4b32bf3f4d6c2416450b : FAILURE in 6m 13s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/31a6b6610fd44ba1b14d17693ab8779a : SUCCESS in 55m 37s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/9bc7e0756a704463b6a07b8e88f39572 : SUCCESS in 53m 26s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/659e7361a0c04fb4beb2783e09201923 : SUCCESS in 47m 12s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/48b50cc826bf4d07b665deea259309b8 : SUCCESS in 1h 11m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01acd23b9f3b48bf91809ff1cfad0014 : SUCCESS in 11m 26s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/5b3041f5da2c4681b29a3bddca14147d : POST_FAILURE in 1h 10m 59s (non-voting)\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/93f243320bdf409cb6cd8de19b8879a8 : SUCCESS in 49m 13s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":1},{"id":"463f5a32fc31b8a6063a2d5bfb9e976d61a15128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 11:56:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db3cd21045917d523101a0b63553f3ef2d5602d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 12:56:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2eb91ba10fc44b79a3b4a02366d7ec6f : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c58ae62594da4b988fe1860660594030 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8507f2d828394944ac0453e09fedd435 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ff4a36b24e8457dba2c0e8efc826f07 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1e964eb8134bb09a966186898683e8 : SUCCESS in 6m 07s\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/4aeecd9285204b20b6ba57c7cd789a1e : SUCCESS in 48m 42s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/a4926a62f0a6468d8d1f08375b5da97f : SUCCESS in 51m 23s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/e634c4db60f743b394c80930df41b2b3 : SUCCESS in 48m 00s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/ba53ebb4f28541338f6398b67db9849f : SUCCESS in 52m 52s\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/fac52ac8861745e886154777ad4825ec : SUCCESS in 49m 57s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":1},{"id":"17f569ccd5dcbab04a53ac6645c7821236f27830","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 12:56:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"13781cb63be23b701a4baeadc02b81774b7411e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 12:58:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bd0e3c3a52a4563b99a44a2d7977c0f : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4b666cbbde8d8835936ad594b6df2a0cbe10529","revisions":{"b4b666cbbde8d8835936ad594b6df2a0cbe10529":{"kind":"REWORK","_number":1,"created":"2020-12-04 07:21:30.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/82/765482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/82/765482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/82/765482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/82/765482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/82/765482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/82/765482/1"}}},"commit":{"parents":[{"commit":"7376520e26e973044f7f986461510defb711cdf5","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/7376520e26e973044f7f986461510defb711cdf5"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-12-04 07:15:39.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-12-04 07:15:39.000000000","tz":480},"subject":"Fix the isolated exception message","message":"Fix the isolated exception message\n\nException\u0027s error message should be put right after the Exception\nwith a bracket, othersize it will be ignored.\n\nChange-Id: I0be28c20ff21805d9a1a52b702cf497df7b9069e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/b4b666cbbde8d8835936ad594b6df2a0cbe10529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/b4b666cbbde8d8835936ad594b6df2a0cbe10529"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23078,"name":"Jeremy Freudberg","email":"jeremyfreudberg@gmail.com","username":"jeremyfreudberg"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
