)]}'
{"id":"openstack%2Fsahara~802478","triplet_id":"openstack%2Fsahara~master~Ib6e09ab83b277704a603c02f0927054478cef8a7","project":"openstack/sahara","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-03-19 14:42:49.000000000","reason":"Change was submitted"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2021-12-17 08:43:56.000000000","reason":"Luigi Toscano replied on the change"},"30077":{"account":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"last_update":"2022-03-19 14:42:49.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-03-19 14:42:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib6e09ab83b277704a603c02f0927054478cef8a7","subject":"Switch from legacy OSA job to integrated OSA job","status":"MERGED","created":"2021-07-27 11:08:38.000000000","updated":"2022-03-19 14:46:34.000000000","submitted":"2022-03-19 14:42:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"802478","meta_rev_id":"65ab219c4d87cf4474cd29bb63fb7c12a718993a","_number":802478,"virtual_id_number":802478,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-19 14:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"recommended":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":1,"date":"2022-03-19 14:46:34.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2022-03-19 13:23:44.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-19 14:46:34.000000000","post_submit":true,"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2022-03-19 13:23:44.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"REVIEWER":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 11:09:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 11:31:42.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-09-06 13:10:10.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-10-26 15:25:35.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-12-15 07:59:31.000000000","updated_by":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"reviewer":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"state":"REVIEWER"},{"updated":"2021-12-16 07:49:50.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"289c38e8a897cbf4e731973e7c8ea87eb8debd5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-27 11:08:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d4b8aaafc73aa62531b4689d4997b0764ada6b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 11:09:02.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"3770337b1ace5dce5eefffe3de62b4bef87e7a13","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-27 11:12:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2faf35020133608337206d27b337b88fba17e57c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 12:57:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edc1a61889ad4401adf9b02420591c28 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/429d13faee0d44ee86ca6749d5accaf9 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95ea2193498543fd84491018b8aea14c : FAILURE in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfe525b841414f49be7bae356ae5a8bd : FAILURE in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2451f4ba9efa4aa5a9ce7a386607f1fc : FAILURE in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b66d0118c7c4606870c28cd6817d58a : FAILURE in 5m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0cacd850ee52458998cf94b97afcac57 : FAILURE in 5m 33s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/ee694379cb1a4220a21ce3494adda214 : FAILURE in 33m 57s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/c402c4ed5b524633b9ff0b3ff2c930e4 : FAILURE in 32m 27s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/bfc4801fbf68492d992536ec213f0fcc : FAILURE in 31m 46s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/a39b920390e24013acad0f47c2fb6dc4 : FAILURE in 31m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2e44299aece4f0983e06dbd7b770eb7 : SUCCESS in 6m 24s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/46b83df9f7df45e586928bc8715792ce : FAILURE in 1h 08m 13s (non-voting)\n- openstack_ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c3a83becc1c4107971772b587c6fc5b : SUCCESS in 1h 38m 04s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/settings","accounts_in_message":[],"_revision_number":1},{"id":"054fed1881d1e843b2de414c2d59d82346dc7c33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-27 13:13:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"037bf42d92736957d9217e335e7870a01a1a45fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 14:39:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c21278a3b91548948bf0da15dd35b300 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9caf59a8703c432995532e9a9e182370 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10767a5c4c504284b2c74aeedde140a0 : FAILURE in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d06f608001445d4877d005c8accc6cf : FAILURE in 8m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dce9d21795e46338f876b83e8dc18a5 : FAILURE in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bc32d50784a47f1b049f77437f83125 : FAILURE in 7m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/58c21f50631f4c7587c13d6cee9b9712 : FAILURE in 7m 52s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/0181091839374aa49d4dc65686a4e52b : FAILURE in 33m 58s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/57b5e7fe8e354c249d8588feb009925c : FAILURE in 41m 22s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/b49f4cab1fcf485995b900ef777aaa2b : FAILURE in 35m 00s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/61fcd38937b040f696cf63a866a58671 : FAILURE in 34m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4bdcda4138b4684a989feb55e5c1d2e : SUCCESS in 7m 02s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/b658e149a16c4b7ab55201b86ca0be94 : FAILURE in 1h 13m 04s (non-voting)\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d358ee240474d9ca86e47b2ee709ef2 : SUCCESS in 1h 15m 46s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/settings","accounts_in_message":[],"_revision_number":2},{"id":"ee0e1ff22dd8d2550a2703dd5d81408c593e3053","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-27 15:29:49.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f7733dc217a31bf8c58ed856eba87dd8f7f0d667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 15:30:21.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"100e294ebf66bce6063d270dd05861b124fc0342","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-27 16:11:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d9a7dd6d8b423d98de8a230a545d983ddfda104b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 17:52:55.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eff7f8229824159b548e6b288a605cc : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/318d16479ebc433ab5b9c86d46faf953 : SUCCESS in 9m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7816b2a6edc14de983ccc3356cb73ae1 : FAILURE in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64069aa0bb454a2282c3a72fb639e36c : FAILURE in 8m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e07fc1b0b5104c1cb1f80ccd68d87254 : FAILURE in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/566a6ba90a45493fbb1a9655b9b8bf5d : FAILURE in 6m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1b4214df78c1488cb376e42c576d5088 : FAILURE in 6m 53s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/b173ffd528e64ad28af34dfe731c3c4c : FAILURE in 30m 41s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/c4fff7e7f3024b56a94ad05ca07bd9e4 : FAILURE in 32m 24s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/c1117159eb114c1fac93b6dbea7d3e42 : FAILURE in 34m 03s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/3501fdb761914a55b026c26bea792bc8 : FAILURE in 33m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3556b5c5643f4f41b5291a3f62161d8c : SUCCESS in 7m 32s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/7bce9910514b4426b1c9e965ea5670bc : FAILURE in 1h 24m 50s (non-voting)\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f404690d837746beb2eb5c1677dcbc98 : RETRY_LIMIT in 17m 17s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/settings","accounts_in_message":[],"_revision_number":3},{"id":"bfc24c4784d3b21b70937c9d8f85d00cb3deaa9a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-27 19:40:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"324391a4b65baa1d7beaa86d2112a2df4a6d84ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 19:40:53.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"2641e6e6ab6c1ec70cd2aef8562bc00f0755ddd1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-27 22:13:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fdad4dc22ece55198005a2ce7ac58a6f140ae1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 22:14:27.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"a5c4b8c005d7edb6ffce532750b6ad07aa662288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-28 08:23:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bca01fb86b056971e3b70b62e005ac64c4d6c4b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 08:24:20.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/sahara on branch master.  The error was:\n\n  Job openstack_ansible-deploy-aio_sahara_metal-ubuntu-focal not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-lower-constraints-jobs\n        - openstack-python3-wallaby-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - check-requirements\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-pylint:\n              voting: false\n          - sahara-tests-scenario\n          - sahara-tests-scenario-v2\n          - sahara-tests-tempest\n          - sahara-tests-tempest-v2\n          - openstack-tox-cover:\n              voting: false\n          - sahara-grenade:\n              voting: false\n          - openstack_ansible-deploy-aio_sahara_metal-ubuntu-focal\n      gate:\n        queue: sahara\n        jobs:\n          - sahara-tests-scenario\n          - sahara-tests-scenario-v2\n          - sahara-tests-tempest\n          - sahara-tests-tempest-v2\n          # - sahara-grenade\n          - openstack_ansible-deploy-aio_sahara_metal-ubuntu-focal\n      experimental:\n        jobs:\n          - sahara-buildimages-ambari\n          - sahara-buildimages-cloudera\n          - sahara-buildimages-mapr\n          - sahara-buildimages-spark\n          - sahara-tests-scenario-multinode-spark\n\n  in \"openstack/sahara/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":4},{"id":"d5a858ea0f2c5d0cc0d1b72c14f9f75024410707","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-28 08:26:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f47f5d2886e5e672e957c48ed7bd7269c02283d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 08:26:56.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/sahara on branch master.  The error was:\n\n  Job openstack-_ansible-deploy-aio_sahara_metal-ubuntu-focal not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-lower-constraints-jobs\n        - openstack-python3-wallaby-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - check-requirements\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-pylint:\n              voting: false\n          - sahara-tests-scenario\n          - sahara-tests-scenario-v2\n          - sahara-tests-tempest\n          - sahara-tests-tempest-v2\n          - openstack-tox-cover:\n              voting: false\n          - sahara-grenade:\n              voting: false\n          - openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal\n      gate:\n        queue: sahara\n        jobs:\n          - sahara-tests-scenario\n          - sahara-tests-scenario-v2\n          - sahara-tests-tempest\n          - sahara-tests-tempest-v2\n          # - sahara-grenade\n          - openstack-_ansible-deploy-aio_sahara_metal-ubuntu-focal\n      experimental:\n        jobs:\n          - sahara-buildimages-ambari\n          - sahara-buildimages-cloudera\n          - sahara-buildimages-mapr\n          - sahara-buildimages-spark\n          - sahara-tests-scenario-multinode-spark\n\n  in \"openstack/sahara/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":5},{"id":"bb15408b0ebca3215dc4e97b1f71122dadf7f405","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-28 08:28:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"29b14b6a56b841a22450543b2681c7cf7a9e45e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 09:55:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6bcd89e89a3470ebe7e1d0e2fcd2c78 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c84772b6620e406a8dfbe552abbb6654 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/872fffd8a62a4e2d8a546116438f7ad9 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad0953d062734a3f8ed2b9627a1ae392 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e000d38ee42c4594b2e40e5e6ae16865 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f311e5ec682945e69a0718002154dba0 : SUCCESS in 8m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/75e0a6fe8eb14efc97656e6c463aa105 : FAILURE in 5m 40s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/7132d3178ca742f1a8c98cee70edcdcb : SUCCESS in 56m 37s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/5c7a969c19e5488eb86a98ced3eecaa9 : SUCCESS in 47m 51s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/1cb788ad973c40d2812c2c50fbdaa6db : SUCCESS in 38m 16s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/c97a66288d3e4b05b198ad33a80ee902 : SUCCESS in 36m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39d4972a46674823ba789d223ce22d07 : SUCCESS in 9m 14s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/ccbb304c2ac340d991fe10b62b7440f9 : FAILURE in 1h 18m 52s (non-voting)\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b8c0a2889154fe7aebde053ffef24ec : SUCCESS in 1h 17m 13s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/settings","accounts_in_message":[],"_revision_number":6},{"id":"a2c4fa163fade44f93a4165ef2fb1365c99784fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-28 10:26:18.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"84275f2df6e3f8991422b7ad09725d40c41ac4c5","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-07-28 11:31:42.000000000","message":"Patch Set 7:\n\nIf you need to change the commit message, please add \n\nStory: 2009081\nTask: 42894","accounts_in_message":[],"_revision_number":7},{"id":"00cde1d260f58e7f93ad256bfc35f2a39fc548fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 11:36:25.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/418fd102da934c36b47f02cdf28edccf : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f92cedb642a4549a589731724db83ef : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ac07308387e4be8b7218ccbc759a101 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c0287e03a1b40dd854c56258f9c6432 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/593732c774fd4a9498f1153bbefa5f22 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e97374c58f0a48739e10f91b18368c33 : SUCCESS in 7m 01s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3e89219b8d5d4b4685d065249af303b7 : FAILURE in 6m 04s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/c5c8d990fe7c4f6c8879e0d5cb313e27 : SUCCESS in 53m 23s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/319420c64acf48db82bf1a34d8c6a03e : SUCCESS in 56m 39s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/62ad1fa6487548aabd8d2902d2ad56b0 : SUCCESS in 46m 50s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/e83b77bd49964d0f9b63a4003d118ceb : SUCCESS in 42m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ae0eacfd3ef4495a676b2b9feb82677 : SUCCESS in 9m 00s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/b3fcedc94c1c43058e6f65d14081f5f8 : FAILURE in 1h 07m 03s (non-voting)\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cfb928f40394725a281e05933329892 : FAILURE in 41m 56s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":7},{"id":"5d021d28d04edee0802c5bde5828bebe84dbd170","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-28 11:58:06.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"1b84e5a4a92b30ae2d00db2a5dbef14df6f3371b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-28 11:58:41.000000000","message":"Patch Set 8:\n\n\u003e Patch Set 7:\n\u003e \n\u003e If you need to change the commit message, please add \n\u003e \n\u003e Story: 2009081\n\u003e Task: 42894\n\nDone - it\u0027s failed on upstream erlang repos being broken, so we can try again.","accounts_in_message":[],"_revision_number":8},{"id":"f0cb0e91419a987cbb71daa888bf137b6c1902a8","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-07-28 12:03:52.000000000","message":"Patch Set 8:\n\nIt seems the sahara tests are still not executed at this stage","accounts_in_message":[],"_revision_number":8},{"id":"b5f852a69ff841cb2075bf4fe43c86a47d923f9c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-28 12:08:29.000000000","message":"Patch Set 8:\n\n\u003e Patch Set 8:\n\u003e \n\u003e It seems the sahara tests are still not executed at this stage\n\nWill get there in the end - I messed up the depends-on :/","accounts_in_message":[],"_revision_number":8},{"id":"3d41ad14923c20e165e62477270ef30b18811f8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 13:29:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82cf9ab86b294cfe96b657a84972d423 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d44995621e494f5bb70d83b153a393d5 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d59bb27e1244f2d995434a35db97b3b : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3168de1df7a946fca5a4c11f02966849 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc6d576fc0a54202af8e7bbae90a8949 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e0e31350f41470c8f5ef70b1d783896 : SUCCESS in 6m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/562a5dc9252a4a07bdf99ddc4f7f84bd : FAILURE in 6m 02s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/5479145fa907433ab886dcdd8aef8e22 : SUCCESS in 53m 36s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/941e30c0b91a48fe9c99e973581b3a13 : SUCCESS in 46m 24s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/76c6d9ac0ce0400c92ba8a5957a504cf : SUCCESS in 43m 40s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/41b3512c9b4f4f2da29bd6893a8e5e72 : SUCCESS in 40m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09ec1a6a585d4924a71163f602eb9ed4 : SUCCESS in 7m 45s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/6e0edaf5c2f24e5482602ef50dfdb8fe : FAILURE in 1h 13m 22s (non-voting)\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6423b9e9807a4b43aec501d793fe7488 : SUCCESS in 1h 23m 48s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":8},{"id":"a490fd1db0e1b4e6ee0c8fedd0661265ec71f8db","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-07-28 13:50:57.000000000","message":"Patch Set 8:\n\nOK, some progress https://780a11778aeb29655ec5-5f07f27cfda9f7663453c94db6894b0a.ssl.cf5.rackcdn.com/802478/8/check/openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal/6423b9e/logs/openstack/aio1-utility/stestr_results.html\n\nHow does that look?","accounts_in_message":[],"_revision_number":8},{"id":"37f359a7ba09fac5785d64e2c4a12ceff03b5f0d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-07-28 16:22:26.000000000","message":"Patch Set 8:\n\n\u003e Patch Set 8:\n\u003e \n\u003e OK, some progress https://780a11778aeb29655ec5-5f07f27cfda9f7663453c94db6894b0a.ssl.cf5.rackcdn.com/802478/8/check/openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal/6423b9e/logs/openstack/aio1-utility/stestr_results.html\n\u003e \n\u003e How does that look?\n\nIt looks fine, thanks! So do you think the sahara-specific job should be defined in the shared repository instead of here?","accounts_in_message":[],"_revision_number":8},{"id":"6878d92e70a5bae0ca8571059b20cb06c8e80bdb","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-08-30 08:21:44.000000000","message":"Patch Set 8:\n\nping (see my question above)","accounts_in_message":[],"_revision_number":8},{"id":"1f1176dc5e78d5f3ce401bb9efda8b7015ac6159","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-06 13:10:10.000000000","message":"Patch Set 8:\n\nWell, yes, I\u0027d say we can define it here instead. But it\u0027s important that job to be parented from openstack-ansible-deploy-aio job.","accounts_in_message":[],"_revision_number":8},{"id":"7aab9f4ddaf3f9166a5ff9d2b28509e770254b32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-06 13:12:07.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"3855e31e3a35c847cdeec124032f4916c54dc87f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 14:33:36.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f16cb95ba4964fceb69ba6f998abe4d7 : SUCCESS in 15m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c481e4fbb2cb4f9a8a6ff5700c277693 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/462308340d0945db99da5d8f25e68b2e : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b72e464df96457aa84b8f311a0550e4 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c0aa63671a7454e9ac48723171536b4 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fda3970398448c885cceec15e96deb9 : SUCCESS in 6m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0f49c971f8745b08ddb88a295eeab71 : FAILURE in 5m 58s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/1466169b2f2943d9993fb6f70a3b79e1 : SUCCESS in 55m 37s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/6fc0aa563154428dbf8785dfc5c3cc9a : SUCCESS in 56m 11s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/df0b70b362134bed987cc5f42bacebfd : SUCCESS in 41m 20s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/5883661fe8e8499ab1a7b366f97b123c : SUCCESS in 36m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ac872df962445bfb291eb8283a33938 : SUCCESS in 7m 59s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/10af5184d0d347a1a5c8bb1da51cb01b : SUCCESS in 1h 15m 17s (non-voting)\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e85ba59293be41728ae26d3760e03459 : SUCCESS in 1h 14m 12s\n\nWarning:\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/resources.sh\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/plugin.sh","accounts_in_message":[],"_revision_number":9},{"id":"6a2f4f414bfd0dae84ce801da533107de0e97b4f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-10-26 15:25:35.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"d3de0991cdf325c71a2af1c01d07c8d308a7c55f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-02 11:03:25.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6c2b6252d78db98602bc276b47df8aeae87df325","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-12-15 07:59:31.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2dd559c2a73db0c86a395cc439ebecaa8fe5b628","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-15 10:02:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5c4052a3073923ed84bd49df8e3db036cb468be5","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-12-15 11:05:34.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"10e176f7774a1bbabac730edac7d333e3b3187f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 11:09:16.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"647b0110c8b0943cdd2bcb2b3d45fd2e123e78fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 12:58:06.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df42d41fd1e14b729e22db74abf516de : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9766cd15067c43d69a40b8655b2c2828 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/616cd2e58361486da0b6dfebcf2cdc65 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b0477ccabe84b5283051894eaedaa3b : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17f43ad3a4064f12b37049f472a37ad3 : SUCCESS in 7m 50s\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/cc79d7a69a2b4a81a201a07c7ee0b1a1 : FAILURE in 51m 35s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/d6051d145bd243db97e485effaeeb6bd : FAILURE in 48m 34s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/7a6245565b004caebd9ffa5b302df8cd : SUCCESS in 38m 35s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/80f6abd6900d471f9f0a0c9fd6903e90 : SUCCESS in 35m 33s\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/817aedbfa5f44a82bcedc608ca4065ce : SUCCESS in 1h 43m 53s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":9},{"id":"ffccea50226a2383efb4acc827d2fda1c44d9085","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-12-16 01:23:46.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c0357a3e7a0ee26c65c1efb2f66e053def02134f","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-12-16 01:33:03.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e8a9a6fc3a8f4dd0a333c0cb45c4bfa0ab95fd67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 02:25:44.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d8bd62d521341e2a033f6f4032d34cb : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc559f6a0ef34ee0a93940e7a1191cb4 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc43f2a7a5324032bfa0944ef0a985f0 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d29fd07218a49469d6ae2b93de77c5d : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc0db3de3d3a46e681c23e7db3d44f71 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1131c41d5b747f2be406ec585141b2c : SUCCESS in 6m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/af1d03d0cc544ef58582bf5753ef768c : FAILURE in 5m 19s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/8de11c5422ac441d9fc571c6c2db1d7b : FAILURE in 40m 08s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/f2e2058d233844dda5d39d8dea78a546 : FAILURE in 48m 42s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/8ebd0a0c03bb460ca9ba703eea404252 : SUCCESS in 32m 40s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/17ae9cedac564b24844fe30ebcb1fc9f : SUCCESS in 35m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfda97008abc42cfac3ee6afd7af9725 : SUCCESS in 7m 23s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/45f36d1588f2490892e34e4c4724731e : SUCCESS in 58m 39s (non-voting)\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/489d9a74e5dc4edc996e86d89d7bcb1e : RETRY_LIMIT in 16m 47s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":9},{"id":"0d0cc6da9b2dd4cc18b1b03c27dc99cc79db6166","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-12-16 07:49:50.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f43627025b3431f451712a1a2ea75b93e3a8b12d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 09:19:25.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9e313905cea428f9433a1589d1d9cf7 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b9637dbb9324bb1a0a55a63c9a7b505 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be012f23415646f08184ec30dd039067 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8f14cec42a2451bbd12642234c9cb1d : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3b2876cab614201a105e81dc239424d : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b7c006365fe4ca1b2eda55b5803935e : SUCCESS in 5m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/089f99d5a0b94fba9f3f2d2cf364a713 : FAILURE in 5m 05s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/9c21eaa350d744e9b1e3a7976be3a169 : FAILURE in 56m 48s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/c8ee315e30f748ada85b1504a18fe084 : FAILURE in 52m 06s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/0fde4f9735d54b23a3363212e4a297dc : SUCCESS in 44m 42s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/18ce8024ea064687baaadafd87009377 : SUCCESS in 41m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3baa78e1e174ce98e78e6c4a5aa8074 : SUCCESS in 7m 14s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/edf9b447257740b38da4a765c6ae6720 : SUCCESS in 1h 20m 57s (non-voting)\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dc0bf04309942ed879deaed7f8b54f2 : RETRY_LIMIT in 13m 34s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":9},{"id":"c72457b3f443e60ad9d47d46374a316212137383","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-12-16 09:52:28.000000000","message":"Patch Set 9: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ff4f6ef21b4878e3dab0ccfe6bac3061d4720396","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-12-16 09:53:13.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8c5bbcb6349cfa6d6c88063bad560fc701d715b5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-16 14:30:09.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"94acd95dce579be60220147ab0a73f09e07d5d2a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-16 14:30:36.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3ce5960add2e5bef809aecc46ad38ad2be152cae","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-12-17 05:39:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"288b60091d8a531e1ca8f7a50641bf6eae8212e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 07:02:35.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2165c42a57d542ed97538eeec6f6b1b4 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37cf2c8aa9b9482792c70533792e8231 : FAILURE in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ac98b752bfa4463a5843ea4454e3423 : FAILURE in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4f0783de6e14440a8f6ca60502c9bae : FAILURE in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43adde533912431594572fd5a5c301f3 : FAILURE in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e46eb73b80e473286393cbc2d53eedb : FAILURE in 4m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d458222e43a14b8f95a4499cfc7094bf : FAILURE in 4m 29s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/1ee4af9048644f80aea7782ee38a8547 : FAILURE in 17m 15s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/fead165c915d4d2fa992494f3e3b6eed : FAILURE in 17m 47s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/30bb225f452b4d51aaee516790b51c92 : FAILURE in 16m 22s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/f16c4074637140c1b95f7f78971ea980 : FAILURE in 8m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f4acb4a1c864d97919965654601acb9 : FAILURE in 5m 31s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/309af41c56114dc6b1e864d8d14985d8 : FAILURE in 17m 34s (non-voting)\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45986c96be9c4e01808f4a47f53e420c : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":9},{"id":"6feeca48511dc59694eee54bca0999312e714031","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-12-17 08:43:56.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e19e28c1a89854d5e90edd224b1470d3250c7e8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2021-12-20 03:24:20.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"d6185204bb1407ca6a0db06acfe517f1d16887a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 04:51:20.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/777512f8861b42fab936b8ff9ea8c535 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de0a0d23495148cb9a0966d1ca278ed0 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77c069b936594df1a5a62ee4276fb771 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4980d295cd141dea54706dee388301b : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10ea0a3e43894a43af53953af3992f78 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7379e223b1404f1fb66b8cc554ec2160 : SUCCESS in 6m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b6f1ad50a2fd4dc5b0e8de8a91b4659d : FAILURE in 5m 30s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/e91c01bc97ff496fa03f228c7de4f8a1 : FAILURE in 45m 56s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/5adf1981a9d84faebcf33627e2df8495 : FAILURE in 55m 16s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/a972f916775a496994ae1b61fb46c22b : SUCCESS in 39m 57s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/5e7fe01714134e0d943685c1647ffcc2 : SUCCESS in 41m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef147877bb68418d929ccac45516d132 : SUCCESS in 7m 46s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/b584742ea2424030a86d5b2482b0ef91 : SUCCESS in 1h 11m 17s (non-voting)\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1804dbda83a341d08e927f500ad87bbe : SUCCESS in 1h 20m 20s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":10},{"id":"9b717442437a84c5072be900d6aa0c99dd944bca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"date":"2022-03-19 00:53:15.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"2d91bb3d529c7b4381d8a51ce0b3307f453a8962","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 02:13:39.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a8c67322c2a4654ad030cd1dc2be00e : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/037029d6169e4fe48acdd69417f03d38 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40bbf9e4c2b34f029e1a104f37498480 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c59a07fbcb09461ea9670b64f329c3c3 : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ceb29541bf50464791fc08832d63e930 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f73cf65bbd414b34aa98a4cd58d4ba40 : SUCCESS in 7m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c20c27d71c524c41b0c79a081362d6d8 : FAILURE in 6m 54s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/3290787358ea4e53bd88419178a08858 : FAILURE in 46m 56s (non-voting)\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/313048a1fa334690865691054aac9401 : FAILURE in 41m 46s (non-voting)\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/e3a81dca3a5c43d6be2727e1fecc2fb0 : SUCCESS in 32m 27s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/5819b3b2d4c74d03b10d64ff564c8d75 : SUCCESS in 32m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/920c74306d3f43b68e99063bff0e4cdb : SUCCESS in 9m 53s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/28745f32c32b4326a71c462e7a4cd7c6 : SUCCESS in 50m 08s (non-voting)\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbd0e095f0ff46f8acb1eeccd5205d22 : SUCCESS in 1h 15m 42s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":11},{"id":"68cd91ef63349a05ffc85d45e898de74fe07513f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-03-19 13:23:44.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"66b27fecdd923fa3b3799a623b18d45d32e7c1e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 13:26:22.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"1dcb6a309351c4da24da8b543044613ce9ec24f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 14:42:49.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/899105dfdc2f48539747d5c6ce9ce557 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa6d64f20d0e433a8bec391f5a0b1b08 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3197a092fe74e60a0139925ff87f544 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f36495a891a54ce5ab36e12ce6e600db : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62e36e2c3b44c8eac02bacc88dd66e3 : SUCCESS in 7m 21s\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/a34ed76662764bffb91e671854f60403 : FAILURE in 54m 28s (non-voting)\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/72bf9bf5eab24bff85b9b45bfd17324b : FAILURE in 55m 09s (non-voting)\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/7a7866b61204404989fa94deeadfc7f8 : SUCCESS in 56m 21s\n- sahara-tests-tempest-v2 https://zuul.opendev.org/t/openstack/build/217f3dddc1c94fbaa7020e31fefc8247 : SUCCESS in 46m 20s\n- openstack-ansible-deploy-aio_sahara_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/840e19c718824b589a250a7c5cf9c1c8 : SUCCESS in 1h 15m 47s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":11},{"id":"b206f87f2ca39730f9d94809e544fbd960618dda","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 14:42:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"81a9053a286c1d13210c552ad8149716bd673024","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 14:46:34.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55826a2ebde1400d95c4344ee84371f3 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"010fdbe76ff3a4f8f47eec3d30e312d346cf21d6","revisions":{"9728441c0fd77dc50099966c55a4265d5ab3661a":{"kind":"REWORK","_number":1,"created":"2021-07-27 11:08:38.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/802478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/78/802478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/78/802478/1"}}},"commit":{"parents":[{"commit":"622af6949b2f9da0fcd3f8d75ab89b79b50f42d4","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/622af6949b2f9da0fcd3f8d75ab89b79b50f42d4"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 11:06:52.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 11:08:34.000000000","tz":60},"subject":"Switch from legacy OSA job to integrated OSA job","message":"Switch from legacy OSA job to integrated OSA job\n\nThis should allow testing of un-merged changes to sahara in the\ncontext of an openstack-ansible deployment\n\nDepends-On: Ie3a9e34c00cfe6279ffe078e464403805e02c790\nChange-Id: Ib6e09ab83b277704a603c02f0927054478cef8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/9728441c0fd77dc50099966c55a4265d5ab3661a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/9728441c0fd77dc50099966c55a4265d5ab3661a"}]},"branch":"refs/heads/master"},"62906a0263bafe1346bc3006fccc32451fa9e4df":{"kind":"REWORK","_number":2,"created":"2021-07-27 13:13:06.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/802478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/78/802478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/78/802478/2"}}},"commit":{"parents":[{"commit":"622af6949b2f9da0fcd3f8d75ab89b79b50f42d4","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/622af6949b2f9da0fcd3f8d75ab89b79b50f42d4"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 11:06:52.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 13:13:02.000000000","tz":0},"subject":"Switch from legacy OSA job to integrated OSA job","message":"Switch from legacy OSA job to integrated OSA job\n\nThis should allow testing of un-merged changes to sahara in the\ncontext of an openstack-ansible deployment\n\nDepends-On: Ie3a9e34c00cfe6279ffe078e464403805e02c790\nChange-Id: Ib6e09ab83b277704a603c02f0927054478cef8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/62906a0263bafe1346bc3006fccc32451fa9e4df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/62906a0263bafe1346bc3006fccc32451fa9e4df"}]},"branch":"refs/heads/master"},"14d63d2e4a24d6c2126602d71a3868cdb610f41f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-07-27 15:29:49.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/802478/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/78/802478/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/78/802478/3"}}},"commit":{"parents":[{"commit":"622af6949b2f9da0fcd3f8d75ab89b79b50f42d4","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/622af6949b2f9da0fcd3f8d75ab89b79b50f42d4"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 11:06:52.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 15:29:44.000000000","tz":0},"subject":"Switch from legacy OSA job to integrated OSA job","message":"Switch from legacy OSA job to integrated OSA job\n\nThis should allow testing of un-merged changes to sahara in the\ncontext of an openstack-ansible deployment\n\nDepends-On: Ie3a9e34c00cfe6279ffe078e464403805e02c790\nDepends-On: I81b55961bce3bb664f4450ef14327c1634a9e8e1\nDepends-On: I1bde4186335cb6638d2c90f5eeee1cf7944e1415\nChange-Id: Ib6e09ab83b277704a603c02f0927054478cef8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/14d63d2e4a24d6c2126602d71a3868cdb610f41f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/14d63d2e4a24d6c2126602d71a3868cdb610f41f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"db4b3d8068abb9eda0ac23c14282fc50ef009277":{"kind":"REWORK","_number":4,"created":"2021-07-28 08:23:38.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/802478/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/78/802478/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/78/802478/4"}}},"commit":{"parents":[{"commit":"0e2004924042f32e716264547d2f05cf8d6130cf","subject":"Drop support for Block Storage API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/0e2004924042f32e716264547d2f05cf8d6130cf"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 11:06:52.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-28 08:23:20.000000000","tz":60},"subject":"Switch from legacy OSA job to integrated OSA job","message":"Switch from legacy OSA job to integrated OSA job\n\nThis should allow testing of un-merged changes to sahara in the\ncontext of an openstack-ansible deployment\n\nDepends-On: Ie3a9e34c00cfe6279ffe078e464403805e02c790\nChange-Id: Ib6e09ab83b277704a603c02f0927054478cef8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/db4b3d8068abb9eda0ac23c14282fc50ef009277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/db4b3d8068abb9eda0ac23c14282fc50ef009277"}]},"branch":"refs/heads/master"},"2222e1ed2c3008369726b4b30cd49f01bce5c780":{"kind":"REWORK","_number":5,"created":"2021-07-28 08:26:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/802478/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/78/802478/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/78/802478/5"}}},"commit":{"parents":[{"commit":"0e2004924042f32e716264547d2f05cf8d6130cf","subject":"Drop support for Block Storage API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/0e2004924042f32e716264547d2f05cf8d6130cf"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 11:06:52.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-28 08:26:02.000000000","tz":60},"subject":"Switch from legacy OSA job to integrated OSA job","message":"Switch from legacy OSA job to integrated OSA job\n\nThis should allow testing of un-merged changes to sahara in the\ncontext of an openstack-ansible deployment\n\nDepends-On: Ie3a9e34c00cfe6279ffe078e464403805e02c790\nChange-Id: Ib6e09ab83b277704a603c02f0927054478cef8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/2222e1ed2c3008369726b4b30cd49f01bce5c780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/2222e1ed2c3008369726b4b30cd49f01bce5c780"}]},"branch":"refs/heads/master"},"7efe96ad77fe0fdba25d0dd8fdedb1edb8fa6507":{"kind":"REWORK","_number":6,"created":"2021-07-28 08:28:38.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/802478/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/78/802478/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/78/802478/6"}}},"commit":{"parents":[{"commit":"0e2004924042f32e716264547d2f05cf8d6130cf","subject":"Drop support for Block Storage API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/0e2004924042f32e716264547d2f05cf8d6130cf"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 11:06:52.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-28 08:28:23.000000000","tz":60},"subject":"Switch from legacy OSA job to integrated OSA job","message":"Switch from legacy OSA job to integrated OSA job\n\nThis should allow testing of un-merged changes to sahara in the\ncontext of an openstack-ansible deployment\n\nDepends-On: Ie3a9e34c00cfe6279ffe078e464403805e02c790\nChange-Id: Ib6e09ab83b277704a603c02f0927054478cef8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/7efe96ad77fe0fdba25d0dd8fdedb1edb8fa6507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/7efe96ad77fe0fdba25d0dd8fdedb1edb8fa6507"}]},"branch":"refs/heads/master"},"86fe4cdb7a40fb4720f2ee8401bf19680c7ef8ef":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2021-07-28 10:26:18.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/802478/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/78/802478/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/78/802478/7"}}},"commit":{"parents":[{"commit":"0e2004924042f32e716264547d2f05cf8d6130cf","subject":"Drop support for Block Storage API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/0e2004924042f32e716264547d2f05cf8d6130cf"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 11:06:52.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-28 10:26:18.000000000","tz":0},"subject":"Switch from legacy OSA job to integrated OSA job","message":"Switch from legacy OSA job to integrated OSA job\n\nThis should allow testing of un-merged changes to sahara in the\ncontext of an openstack-ansible deployment\n\nDepends-On: I1bde4186335cb6638d2c90f5eeee1cf7944e1415\nDepends-On: I81b55961bce3bb664f4450ef14327c1634a9e8e1\nDepends-On: Ie3a9e34c00cfe6279ffe078e464403805e02c790\nChange-Id: Ib6e09ab83b277704a603c02f0927054478cef8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/86fe4cdb7a40fb4720f2ee8401bf19680c7ef8ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/86fe4cdb7a40fb4720f2ee8401bf19680c7ef8ef"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8cb5668a7b0a6236c98187f9d5d37b994b436df9":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-07-28 11:58:06.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/802478/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/78/802478/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/78/802478/8"}}},"commit":{"parents":[{"commit":"0e2004924042f32e716264547d2f05cf8d6130cf","subject":"Drop support for Block Storage API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/0e2004924042f32e716264547d2f05cf8d6130cf"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 11:06:52.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-28 11:58:06.000000000","tz":0},"subject":"Switch from legacy OSA job to integrated OSA job","message":"Switch from legacy OSA job to integrated OSA job\n\nThis should allow testing of un-merged changes to sahara in the\ncontext of an openstack-ansible deployment\n\nStory: 2009081\nTask: 42894\n\nDepends-On: I1bde4186335cb6638d2c90f5eeee1cf7944e1415\nDepends-On: I81b55961bce3bb664f4450ef14327c1634a9e8e1\nDepends-On: Ie3a9e34c00cfe6279ffe078e464403805e02c790\nChange-Id: Ib6e09ab83b277704a603c02f0927054478cef8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/8cb5668a7b0a6236c98187f9d5d37b994b436df9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/8cb5668a7b0a6236c98187f9d5d37b994b436df9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b129dd784848acdcf97716a35d6a718a58659c5a":{"kind":"REWORK","_number":9,"created":"2021-09-06 13:12:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/78/802478/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/78/802478/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/78/802478/9"}}},"commit":{"parents":[{"commit":"0e2004924042f32e716264547d2f05cf8d6130cf","subject":"Drop support for Block Storage API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/0e2004924042f32e716264547d2f05cf8d6130cf"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 11:06:52.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-06 13:12:02.000000000","tz":0},"subject":"Switch from legacy OSA job to integrated OSA job","message":"Switch from legacy OSA job to integrated OSA job\n\nThis should allow testing of un-merged changes to sahara in the\ncontext of an openstack-ansible deployment\n\nStory: 2009081\nTask: 42894\n\nDepends-On: I1bde4186335cb6638d2c90f5eeee1cf7944e1415\nDepends-On: I123d32a0967d5196723f57a40ab5acfa7a53a440\nChange-Id: Ib6e09ab83b277704a603c02f0927054478cef8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/b129dd784848acdcf97716a35d6a718a58659c5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/b129dd784848acdcf97716a35d6a718a58659c5a"}]},"branch":"refs/heads/master"},"a4baa507d2958b2b3fac9d79ed759b5aaf7a7b62":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-12-20 03:24:20.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/78/802478/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/78/802478/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/78/802478/10"}}},"commit":{"parents":[{"commit":"804ea3776304efa7d9f202b721bd8889ed98cd39","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/804ea3776304efa7d9f202b721bd8889ed98cd39"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 11:06:52.000000000","tz":60},"committer":{"name":"Qiu Fossen","email":"qiujunting@inspur.com","date":"2021-12-20 03:24:20.000000000","tz":0},"subject":"Switch from legacy OSA job to integrated OSA job","message":"Switch from legacy OSA job to integrated OSA job\n\nThis should allow testing of un-merged changes to sahara in the\ncontext of an openstack-ansible deployment\n\nStory: 2009081\nTask: 42894\n\nDepends-On: I1bde4186335cb6638d2c90f5eeee1cf7944e1415\nDepends-On: I123d32a0967d5196723f57a40ab5acfa7a53a440\nChange-Id: Ib6e09ab83b277704a603c02f0927054478cef8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/a4baa507d2958b2b3fac9d79ed759b5aaf7a7b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/a4baa507d2958b2b3fac9d79ed759b5aaf7a7b62"}]},"branch":"refs/heads/master","description":"Rebase"},"010fdbe76ff3a4f8f47eec3d30e312d346cf21d6":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-03-19 00:53:15.000000000","uploader":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"},"ref":"refs/changes/78/802478/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/78/802478/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/78/802478/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/78/802478/11"}}},"commit":{"parents":[{"commit":"b269b3d1d08ce5c9163f32f2a573abad32bd521f","subject":"Merge \"Remove reference to deprecated RequestContext.user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/b269b3d1d08ce5c9163f32f2a573abad32bd521f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-07-27 11:06:52.000000000","tz":60},"committer":{"name":"Qiu Fossen","email":"qiujunting@inspur.com","date":"2022-03-19 00:53:15.000000000","tz":0},"subject":"Switch from legacy OSA job to integrated OSA job","message":"Switch from legacy OSA job to integrated OSA job\n\nThis should allow testing of un-merged changes to sahara in the\ncontext of an openstack-ansible deployment\n\nStory: 2009081\nTask: 42894\n\nDepends-On: I1bde4186335cb6638d2c90f5eeee1cf7944e1415\nDepends-On: I123d32a0967d5196723f57a40ab5acfa7a53a440\nChange-Id: Ib6e09ab83b277704a603c02f0927054478cef8a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/010fdbe76ff3a4f8f47eec3d30e312d346cf21d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/010fdbe76ff3a4f8f47eec3d30e312d346cf21d6"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30077,"name":"Qiu Fossen","display_name":"fossnqiu","email":"qiujunting@inspur.com","username":"qiufossen"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
