)]}'
{"id":"openstack%2Fsahara~838017","triplet_id":"openstack%2Fsahara~stable%2Fvictoria~I7a4071e52f68fea7d07bad690f8590cbb3fb4fa4","project":"openstack/sahara","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-04-21 19:34:06.000000000","reason":"Change was submitted"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2022-04-21 10:37:29.000000000","reason":"Luigi Toscano replied on the change"}},"hashtags":[],"change_id":"I7a4071e52f68fea7d07bad690f8590cbb3fb4fa4","subject":"Drop tripleo centos-7 scen3 job definition - not used","status":"MERGED","created":"2022-04-15 11:24:47.000000000","updated":"2022-04-21 19:36:43.000000000","submitted":"2022-04-21 19:34:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838017","cherry_pick_of_change":838016,"cherry_pick_of_patch_set":1,"meta_rev_id":"4a2fbca26b63cae8e3b2509591645ac723f5173c","_number":838017,"virtual_id_number":838017,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-21 19:34:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-04-21 10:37:29.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2022-04-21 10:37:29.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-15 13:37:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-20 20:20:07.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2022-04-21 10:37:29.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"}],"messages":[{"id":"19a371bfebd27130a8f2c7700107c0c997c1ad05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-15 11:24:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"388b79a11e583bab745c4d67ca7f9d2a56b93dc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 13:37:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e71e344c49f4cfeb4d5b2a94f9e59cb : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea42d03d6e474788be16ce6a2d3eeb90 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/893ebdf87afd4422956da60c2bd5eaea : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d77315fbc9b6481faf9382d227a57400 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23a2fedb65f24131a806ca57b7ddfc4d : SUCCESS in 5m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e71259f42a8c4a84a7c4065127f7bc0d : FAILURE in 5m 30s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/99acb425d4dc46f196905c0c98e45a79 : SUCCESS in 1h 01m 56s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/f9728d1f8b2a4e54bdd63d0dcfbcf350 : TIMED_OUT in 2h 03m 22s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/8a75e0b9517a4cf6b54cfac9bcafe172 : SUCCESS in 43m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/807e633743ec4d98a42a2d0893c5157f : SUCCESS in 7m 10s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/f66bb8cd58cc4eefbb7cb01d17254aa4 : FAILURE in 16m 42s\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/8e1fd55622804ff3a4e4eb55df987025 : FAILURE in 13m 36s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":1},{"id":"5c703125b8b237a77e7cd6f7a82ab57c42d97d6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-20 10:24:46.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"712d08d24b8d4bb3a705a7e25aef381976113837","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-20 10:25:09.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"50582734faa1389a448585b1989c7f5491294181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 11:35:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd3362e02fc541588c5e72975073e2b7 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b5005cd58204e2d8a1fbea1a540a192 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/901d8b729409494f8e5fe92cfb87f4f5 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afaee349125e4ae0afe8e876fbc38604 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a67950d9054216bfbed9888ebf170f : SUCCESS in 6m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a03cdc94b39643a7bddead9bbddde9c1 : FAILURE in 6m 59s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/108749e7998a4a48bcb97d9504636ae6 : SUCCESS in 47m 15s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/c9b2a0b73a424597bef2af17134e60fb : SUCCESS in 50m 03s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/7919a8fdc6144107bbfe7d7dc31847d3 : SUCCESS in 45m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e307f962ef7e47baa580d9f4306050a8 : SUCCESS in 8m 18s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/ed1e6a20a058491ca0d47e098d230c89 : SUCCESS in 1h 07m 14s\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/16ea59359fad41a497d461366b3131d6 : FAILURE in 15m 50s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":3},{"id":"3cddf211ea5974fb65c92c148dd9ec52594ac1e2","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-04-20 20:20:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f7c6d629ac4bd7a3d5cf1734c722914e0d380e7","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-21 08:40:48.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7ecd4dede2d2f889b0384af6e93e7373afcb32bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-21 08:40:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"affadb64f8494eb82b96bfb98f6dda7994294a19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 09:57:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af38e0241bb840159cb90e7af7bdce91 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe854f74e111421381fe47497ec77a4c : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee7fd737d4d54f4a9dc14544c38d9ac1 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a09ac0d25d2a4935b865beb3149f82d8 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b43e97d511e04bff89ddaf58354f30ff : SUCCESS in 6m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bd311d126e5f4a70859d790ed0b66cb3 : FAILURE in 5m 31s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/7e1234c719464d1eb1ebd677330d6d5f : SUCCESS in 58m 39s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/6c50234cf0834c768b7d320d8694b887 : SUCCESS in 57m 09s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/a2b69c8fcd8d4cb79d64906bd34cd014 : SUCCESS in 48m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e4c040bbb4d4177b367e58ba6349501 : SUCCESS in 9m 00s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/afe3782461eb42efa0863a1dfe7a364c : SUCCESS in 1h 12m 56s\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/54731ca96f1e40e4ae43ca09a416bbfd : FAILURE in 46m 34s (non-voting)\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":4},{"id":"73a6619460b68eb99074b093660bac20106c7e24","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-04-21 10:37:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"820f67cc05e2943276d1bc2595603aae0d1849c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 10:38:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ed8536e5dc0f677a3bb2259bc196351076c99247","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 12:46:24.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a75846f98ec84be7a8df7893d4d2780a : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b5510285c4a440e88b72b9dbefe30f5 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66d375e6687d4451930c0f1d11f90e74 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a64309627f0d42c2a7223a36f43d5760 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f7f18e6c5f4e2998b0fc16120e3592 : SUCCESS in 6m 55s\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/edbe032a959345a48940744269dcf068 : SUCCESS in 56m 24s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/bc4f5a9512bc4361ace8ba792e7a9ef9 : TIMED_OUT in 2h 03m 21s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/04ff9f9f79074c1f8d171ba41f111cdf : SUCCESS in 52m 35s\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/950b01e66a7b4362a6ab30db851a97df : SUCCESS in 1h 15m 47s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":4},{"id":"9acdc97a96a732e3f9319f18c9d6bc38a9254a90","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-04-21 13:13:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4aed10b0e62be8a75167a167c62cf6f348e8cca3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 14:26:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f137d9557a04d15b4a16f96b93885d5 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e468aeda58b4e399f5d2cc15a2f0189 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe2cc7a1b29641ff9007a5166a474ced : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d4c8dc8d28f4eabbfa73deadfc20fe6 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e63bdc75a04d5aad3cd4c26f10ba9f : SUCCESS in 6m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38982a86d31146dca0c5d8fc557357a6 : FAILURE in 6m 08s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/aba1ea652a7e48c5b75a408e8b4e5937 : SUCCESS in 51m 16s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/efc2c8291de84e5aa1df4fcc5eef88a6 : SUCCESS in 51m 56s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/c10da0927218428fb4681f0cbc267f07 : SUCCESS in 41m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff00d56884444acbbdf0a59b11be1d64 : SUCCESS in 9m 44s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/e7fbd8243a9247048e15a493b2f68a47 : SUCCESS in 1h 07m 51s\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/4e7bad2417a548699b60f4115c068668 : FAILURE in 49m 06s (non-voting)\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":4},{"id":"77fee5884b063e493c179c0581f857cff6a1bc81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 14:28:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ff97d35480b0b73a73f74c23f8882822b37c0464","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:38:25.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d420ebc54c17445cac2c7d2cf0df0926 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fd2adacbc3e436a92e48b7fd0b05ae2 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f2f0de517af4bf9a1abde5aa9cd76b8 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46901e56007145a1941f8cef5268fffa : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64512f58a3514321940ae5c9cdeaf46e : SUCCESS in 7m 42s\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/d6fb4c9738f34b0babac1c5d3554073a : SUCCESS in 1h 05m 26s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/f0151aea394c42bc9fe7d0dc9e845451 : TIMED_OUT in 2h 03m 52s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/4524b86a01f34752b313a47d6e2859f8 : SUCCESS in 54m 44s\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/ab26aeba37544227abe292c229b74ca5 : SUCCESS in 1h 16m 23s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":4},{"id":"09d682ddc95ddcc8c7719bdfe71a73f91726b881","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-04-21 16:54:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b9fe238d3b3d3e2f15956363b367b10958400070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 18:20:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25c659ec54fc4ab5acff72e9da35b7e5 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c191bdd2f4484ff2850edc25bca8476e : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9652ce251ba044649d18dc09328786b7 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14dc00c05e064083923a83a1e3df4aca : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3b849aac5545d885f15acf9435cf5d : SUCCESS in 6m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4422333dff1d4198a764d6254a5c60dc : FAILURE in 6m 23s (non-voting)\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/2157974a48c24d2c8ba4e72ba219e7db : SUCCESS in 56m 57s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/9e342625aad540efb531e827bd98a7fd : SUCCESS in 1h 00m 04s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/0cb5dcc5c2b8453fa8bd65d3a14999f3 : SUCCESS in 47m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1275b3ad992646bd8bb928a1af84de95 : SUCCESS in 8m 06s (non-voting)\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/4941889a0556414ba48b8436fafa2ecb : SUCCESS in 1h 23m 15s\n- sahara-openstack-ansible-functional https://zuul.opendev.org/t/openstack/build/8fba1631044947f1834f3525b6a0c517 : FAILURE in 45m 57s (non-voting)\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":4},{"id":"914f1d0a585c533cc43cc24537394322e688a932","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 18:22:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"97b0abd07efa0b05c4b335e64858cd5129fa5293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 19:34:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98c6f147207c4b0a8b411285f174cd0f : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bae75ae74b34cd2aaa03619a645bdb2 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a627a456cb5d4cfb88b4f8639263de1c : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/386a0db650e94a478e211db0614cbed7 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a0be4cd456c4afaae58aba997544837 : SUCCESS in 6m 15s\n- sahara-tests-scenario https://zuul.opendev.org/t/openstack/build/8c71ddc46a254e17808111f4c017ccda : SUCCESS in 58m 10s\n- sahara-tests-scenario-v2 https://zuul.opendev.org/t/openstack/build/67c81afe64b14507809c54ed2aba1574 : SUCCESS in 55m 26s\n- sahara-tests-tempest https://zuul.opendev.org/t/openstack/build/f3b3d0d0856a4a448424d0a0543b1e0d : SUCCESS in 46m 12s\n- sahara-grenade https://zuul.opendev.org/t/openstack/build/d8b2308ae21a42578ecd181ba776fcc5 : SUCCESS in 1h 11m 42s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/settings\n  Comments left for invalid file devstack/upgrade/from-mitaka/upgrade-sahara\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":4},{"id":"e939ce086d59a85243c729e04c7d903c73d85938","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 19:34:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"4a2fbca26b63cae8e3b2509591645ac723f5173c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 19:36:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1077127e44f349c2946a94f760cd62c2 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4175be12a64c23f3dd1268d3cbd7d68a3e42a825","revisions":{"b151215ab96e6960871230140f6e74f0bc36c847":{"kind":"REWORK","_number":1,"created":"2022-04-15 11:24:47.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/17/838017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/17/838017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/17/838017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/17/838017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/17/838017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/17/838017/1"}}},"commit":{"parents":[{"commit":"44ac39ad4d58c41dfa591e4873fcfc84102850d6","subject":"Preinstall requirements for docs build and fix lower constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/44ac39ad4d58c41dfa591e4873fcfc84102850d6"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-15 11:06:12.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-15 11:24:47.000000000","tz":0},"subject":"Move replace centos-7 tripleo scen3 job with centos-8","message":"Move replace centos-7 tripleo scen3 job with centos-8\n\nThis is needed by [1] where this job is being removed. Note\nthat this job as defined will only run for stable/stein [2].\nAlso note that this job doesn\u0027t appear to be running anywhere\nit is not wired up in this repo [3] or anywhere else.\n\n[1] https://review.opendev.org/c/openstack/tripleo-ci/+/838110/2#message-2efaca7aaf100874fd6996f18f0e74ae942149c0\n[2] https://opendev.org/openstack/tripleo-ci/src/commit/4daed1b14c582b442f64ec45b1ddba5b58e03986/zuul.d/standalone-jobs.yaml#L476\n[3] https://codesearch.opendev.org/?q\u003dsahara-tripleo-ci-centos-7-scenario003-standalone\u0026i\u003dnope\u0026literal\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d\n\nChange-Id: I7a4071e52f68fea7d07bad690f8590cbb3fb4fa4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/b151215ab96e6960871230140f6e74f0bc36c847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/b151215ab96e6960871230140f6e74f0bc36c847"}]},"branch":"refs/heads/stable/victoria"},"cced31f9de71580570e681eeb37f5fb0a20a7df3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-20 10:24:46.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/17/838017/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/17/838017/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/17/838017/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/17/838017/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/17/838017/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/17/838017/2"}}},"commit":{"parents":[{"commit":"44ac39ad4d58c41dfa591e4873fcfc84102850d6","subject":"Preinstall requirements for docs build and fix lower constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/44ac39ad4d58c41dfa591e4873fcfc84102850d6"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-15 11:06:12.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-20 10:24:46.000000000","tz":0},"subject":"Drop tripleo centos-7 scen3 job definition - not used","message":"Drop tripleo centos-7 scen3 job definition - not used\n\nThis is needed by [1] where this job is being removed. Note\nthat this job as defined will only run for stable/stein [2].\nAlso note that this job doesn\u0027t appear to be running anywhere\nit is not wired up in this repo [3] or anywhere else.\n\n[1] https://review.opendev.org/c/openstack/tripleo-ci/+/838110/2#message-2efaca7aaf100874fd6996f18f0e74ae942149c0\n[2] https://opendev.org/openstack/tripleo-ci/src/commit/4daed1b14c582b442f64ec45b1ddba5b58e03986/zuul.d/standalone-jobs.yaml#L476\n[3] https://codesearch.opendev.org/?q\u003dsahara-tripleo-ci-centos-7-scenario003-standalone\u0026i\u003dnope\u0026literal\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d\n\n(cherry-picked from commit af0faafaad26a3085d81c3ed17f58727b33444c4)\n\nChange-Id: I7a4071e52f68fea7d07bad690f8590cbb3fb4fa4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/cced31f9de71580570e681eeb37f5fb0a20a7df3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/cced31f9de71580570e681eeb37f5fb0a20a7df3"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"},"b1aad097244ccc2c136d85706217a05a3dd5116c":{"kind":"REWORK","_number":3,"created":"2022-04-20 10:25:09.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/17/838017/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/17/838017/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/17/838017/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/17/838017/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/17/838017/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/17/838017/3"}}},"commit":{"parents":[{"commit":"44ac39ad4d58c41dfa591e4873fcfc84102850d6","subject":"Preinstall requirements for docs build and fix lower constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/44ac39ad4d58c41dfa591e4873fcfc84102850d6"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-15 11:06:12.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-20 10:25:09.000000000","tz":0},"subject":"Drop tripleo centos-7 scen3 job definition - not used","message":"Drop tripleo centos-7 scen3 job definition - not used\n\nThis is needed by [1] where this job is being removed. Note\nthat this job as defined will only run for stable/stein [2].\nAlso note that this job doesn\u0027t appear to be running anywhere\nit is not wired up in this repo [3] or anywhere else.\n\n[1] https://review.opendev.org/c/openstack/tripleo-ci/+/838110/2#message-2efaca7aaf100874fd6996f18f0e74ae942149c0\n[2] https://opendev.org/openstack/tripleo-ci/src/commit/4daed1b14c582b442f64ec45b1ddba5b58e03986/zuul.d/standalone-jobs.yaml#L476\n[3] https://codesearch.opendev.org/?q\u003dsahara-tripleo-ci-centos-7-scenario003-standalone\u0026i\u003dnope\u0026literal\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d\n\n(cherry-picked from commit af0faafaad26a3085d81c3ed17f58727b33444c4)\n\nChange-Id: I7a4071e52f68fea7d07bad690f8590cbb3fb4fa4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/b1aad097244ccc2c136d85706217a05a3dd5116c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/b1aad097244ccc2c136d85706217a05a3dd5116c"}]},"branch":"refs/heads/stable/victoria"},"4175be12a64c23f3dd1268d3cbd7d68a3e42a825":{"kind":"REWORK","_number":4,"created":"2022-04-21 08:40:57.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/17/838017/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/17/838017/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/17/838017/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/17/838017/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/17/838017/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/17/838017/4"}}},"commit":{"parents":[{"commit":"95a84e03586d4f191347c752bcbd6bf25c78411b","subject":"Merge \"Let the code in the json file be displayed\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/95a84e03586d4f191347c752bcbd6bf25c78411b"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-15 11:06:12.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-21 08:40:28.000000000","tz":180},"subject":"Drop tripleo centos-7 scen3 job definition - not used","message":"Drop tripleo centos-7 scen3 job definition - not used\n\nThis is needed by [1] where this job is being removed. Note\nthat this job as defined will only run for stable/stein [2].\nAlso note that this job doesn\u0027t appear to be running anywhere\nit is not wired up in this repo [3] or anywhere else.\n\nCherrypick is not clean, it makes sahara-openstack-ansible-functional\nnon voting as requested in review comments [4]. This part is\nspecific to Victoria and Ussuri.\n\n[1] https://review.opendev.org/c/openstack/tripleo-ci/+/838110/2#message-2efaca7aaf100874fd6996f18f0e74ae942149c0\n[2] https://opendev.org/openstack/tripleo-ci/src/commit/4daed1b14c582b442f64ec45b1ddba5b58e03986/zuul.d/standalone-jobs.yaml#L476\n[3] https://codesearch.opendev.org/?q\u003dsahara-tripleo-ci-centos-7-scenario003-standalone\u0026i\u003dnope\u0026literal\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d\n[4] https://review.opendev.org/c/openstack/sahara/+/838017/3#message-3cddf211ea5974fb65c92c148dd9ec52594ac1e2\n\n(cherry-picked from commit af0faafaad26a3085d81c3ed17f58727b33444c4)\n\nChange-Id: I7a4071e52f68fea7d07bad690f8590cbb3fb4fa4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/4175be12a64c23f3dd1268d3cbd7d68a3e42a825"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/4175be12a64c23f3dd1268d3cbd7d68a3e42a825"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
