)]}'
{"id":"openstack%2Fsahara~890872","triplet_id":"openstack%2Fsahara~master~Ia24a706d93fa0d717978b127d53135e225f05da3","project":"openstack/sahara","branch":"master","attention_set":{},"removed_from_attention_set":{"36096":{"account":{"_account_id":36096,"name":"guolimin","email":"guolimin.lc@inspur.com","username":"guolimin.lc"},"last_update":"2023-08-09 06:09:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia24a706d93fa0d717978b127d53135e225f05da3","subject":"Remove redundant \"\u003d\" to imporve doc format","status":"MERGED","created":"2023-08-09 05:49:36.000000000","updated":"2023-08-09 06:10:56.000000000","submitted":"2023-08-09 06:09:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890872","meta_rev_id":"946522878ba8efe9f1aeed9a5e7034d017d8a0fd","_number":890872,"virtual_id_number":890872,"owner":{"_account_id":36096,"name":"guolimin","email":"guolimin.lc@inspur.com","username":"guolimin.lc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-09 06:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-09 05:58:48.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-09 05:58:48.000000000","_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 05:57:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-09 05:58:48.000000000","updated_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"reviewer":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"state":"REVIEWER"}],"messages":[{"id":"76176b2f82a7b76a2e2ce5bb59797ffbbb5492c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36096,"name":"guolimin","email":"guolimin.lc@inspur.com","username":"guolimin.lc"},"date":"2023-08-09 05:49:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7431badc5dde8a72850e14c5eb249233e37a1d22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 05:57:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/210a92c147a34564b7cb1bc182b2cc62\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b05f61f3e2cb4403b889ef3f00bace19 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b00b3d8bd7144da89816342580ae326 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/467eba8392184715bb78ba9b31bcaa0b : SUCCESS in 6m 00s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":1},{"id":"3869b3e4941e6e5a916fff4b14afe95f8e1830d3","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2023-08-09 05:58:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"976fd91e0d7ea9b8e5f1f809d489a7c44df49d17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 05:59:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37fc974ea25e2a3eaeae15c07df95a41e4074811","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:09:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3be8d3e298fc4dbb890e89d634dc8587\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c633a7d1f0a44d95a52fc3b0d4ed4f1e : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e447c9fae9438ebd0f58dcd5b1faad : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f12b7381148645a5a11d803d8eac62be : SUCCESS in 5m 09s\n\nWarning:\n  Comments left for invalid file devstack/plugin.sh\n  Comments left for invalid file devstack/upgrade/resources.sh","accounts_in_message":[],"_revision_number":1},{"id":"02a68b48542c544770eaeceec2a08162880dca89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:09:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"946522878ba8efe9f1aeed9a5e7034d017d8a0fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:10:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01b3f6bb117e451a84cdab6a81dcd356\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c7f162416040038124d6b784a41e47 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7165838e23784e5792ffc30986b50f5d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a806a536b623b4ce1a345d5718505a5f04c987f4","revisions":{"a806a536b623b4ce1a345d5718505a5f04c987f4":{"kind":"REWORK","_number":1,"created":"2023-08-09 05:49:36.000000000","uploader":{"_account_id":36096,"name":"guolimin","email":"guolimin.lc@inspur.com","username":"guolimin.lc"},"ref":"refs/changes/72/890872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sahara","ref":"refs/changes/72/890872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sahara refs/changes/72/890872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sahara refs/changes/72/890872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sahara refs/changes/72/890872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sahara refs/changes/72/890872/1"}}},"commit":{"parents":[{"commit":"c53639d3b030d10d35a94b87995b1addddba8494","subject":"Updated links in documentation to be currect links","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/c53639d3b030d10d35a94b87995b1addddba8494"}]}],"author":{"name":"guolimin.lc","email":"guolimin.lc@inspur.com","date":"2023-08-09 05:48:53.000000000","tz":480},"committer":{"name":"guolimin.lc","email":"guolimin.lc@inspur.com","date":"2023-08-09 05:48:53.000000000","tz":480},"subject":"Remove redundant \"\u003d\" to imporve doc format","message":"Remove redundant \"\u003d\" to imporve doc format\n\nChange-Id: Ia24a706d93fa0d717978b127d53135e225f05da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/a806a536b623b4ce1a345d5718505a5f04c987f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sahara/commit/a806a536b623b4ce1a345d5718505a5f04c987f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
