)]}'
{"id":"openstack%2Fsearchlight-ui~669502","triplet_id":"openstack%2Fsearchlight-ui~master~I15c29e5f8a8fbf30e0ec113e41be86184b3dcbff","project":"openstack/searchlight-ui","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I15c29e5f8a8fbf30e0ec113e41be86184b3dcbff","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 20:07:25.000000000","updated":"2019-07-06 03:27:22.000000000","submitted":"2019-07-06 03:27:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669502-1562383642467-6efaa91b","meta_rev_id":"5370222e5b126eedc54b6d3ba6f354caaa9d5e74","_number":669502,"virtual_id_number":669502,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-07-06 03:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2019-07-06 03:16:49.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2019-07-06 03:16:49.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-06 03:16:49.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-07-06 03:27:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e455ccdf11291734f5fd3a88422fba71470f36e","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 20:07:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b1a9aa15ae638ee011f112f52b2181a67098af7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 20:20:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/02/669502/1/check/nodejs10-npm-run-lint/0f810e2/ : SUCCESS in 4m 47s\n- nodejs10-npm-run-test http://logs.openstack.org/02/669502/1/check/nodejs10-npm-run-test/44d61ea/ : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/669502/1/check/openstack-tox-lower-constraints/00319ba/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/02/669502/1/check/openstack-tox-pep8/973ced5/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/02/669502/1/check/openstack-tox-py27/8112e3c/ : SUCCESS in 4m 56s\n- openstack-tox-py36 http://logs.openstack.org/02/669502/1/check/openstack-tox-py36/20e14dd/ : SUCCESS in 3m 48s\n- openstack-tox-py37 http://logs.openstack.org/02/669502/1/check/openstack-tox-py37/ab2e218/ : SUCCESS in 4m 26s\n- openstack-tox-docs http://logs.openstack.org/02/669502/1/check/openstack-tox-docs/a350786/html/ : SUCCESS in 3m 34s\n- build-openstack-releasenotes http://logs.openstack.org/02/669502/1/check/build-openstack-releasenotes/7d752e2/html/ : SUCCESS in 3m 37s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/02/669502/1/check/horizon-openstack-tox-python3-django111/60e8dbf/ : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"186c50b9613b4df66880e6e655af958d443380da","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-07-06 03:16:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"7a6024eecc6fd671ab5076d3391aa22ce96fddde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 03:17:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"759b93a566d9aa82b62ad698433e41d08fa2bb60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 03:27:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"86789076407c07e82069bd36547d4ff098d96312","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 03:27:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/02/669502/1/gate/nodejs10-npm-run-lint/120c140/ : SUCCESS in 6m 32s\n- nodejs10-npm-run-test http://logs.openstack.org/02/669502/1/gate/nodejs10-npm-run-test/43825fb/ : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/669502/1/gate/openstack-tox-lower-constraints/ff6b29b/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/02/669502/1/gate/openstack-tox-pep8/3a6befe/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/02/669502/1/gate/openstack-tox-py27/0b08540/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/02/669502/1/gate/openstack-tox-py36/85945f5/ : SUCCESS in 5m 02s\n- openstack-tox-py37 http://logs.openstack.org/02/669502/1/gate/openstack-tox-py37/3bdb4aa/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/02/669502/1/gate/openstack-tox-docs/f33d808/html/ : SUCCESS in 3m 27s\n- build-openstack-releasenotes http://logs.openstack.org/02/669502/1/gate/build-openstack-releasenotes/21058c1/html/ : SUCCESS in 4m 26s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/02/669502/1/gate/horizon-openstack-tox-python3-django111/524eec6/ : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3722f834f1f1421e85c8c282b4234f0a07e32fb","revisions":{"e3722f834f1f1421e85c8c282b4234f0a07e32fb":{"kind":"REWORK","_number":1,"created":"2019-07-05 20:07:25.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/02/669502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/02/669502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/02/669502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/02/669502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/02/669502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/02/669502/1"}}},"commit":{"parents":[{"commit":"07766fc29144e87926bd7414b230a59b5bf9f210","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/07766fc29144e87926bd7414b230a59b5bf9f210"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 16:00:58.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 20:07:24.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I15c29e5f8a8fbf30e0ec113e41be86184b3dcbff\nStory: #2005924\nTask: #34245\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/e3722f834f1f1421e85c8c282b4234f0a07e32fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/e3722f834f1f1421e85c8c282b4234f0a07e32fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
