)]}'
{"id":"openstack%2Fsearchlight-ui~690917","triplet_id":"openstack%2Fsearchlight-ui~master~Ia259132f4c102468e15f51900c786634d5f45b95","project":"openstack/searchlight-ui","branch":"master","hashtags":[],"change_id":"Ia259132f4c102468e15f51900c786634d5f45b95","subject":"Add PDF generation support for Searchlight-UI","status":"MERGED","created":"2019-10-24 10:38:48.000000000","updated":"2019-10-24 13:27:07.000000000","submitted":"2019-10-24 13:25:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690917-1571923539590-fe29816b","meta_rev_id":"c9249e47ebe6bbff9f6fa871e7b63d63d13dc825","_number":690917,"virtual_id_number":690917,"owner":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 13:25:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2019-10-24 13:13:40.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-24 10:55:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2019-10-24 13:13:40.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 13:13:40.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-10-24 13:25:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"599298fc4d5b24d6d3d51ca59d5e078b980b6836","author":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"date":"2019-10-24 10:38:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b326d67bc1628f53e20bf457d8104c05a1756ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 10:53:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3ff01ab3e9704dbd8cb1369260f77294 : SUCCESS in 2m 19s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/9da2e830c9ba41c4b686eaeba9429e4f : SUCCESS in 4m 58s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/f16fa2ec672443cd9c667333c611248a : SUCCESS in 5m 20s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/251f4410368c4bdcab9f4600c510d69f : SUCCESS in 4m 56s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/950d841065cf4d8980fee988786ed687 : SUCCESS in 8m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1083c9eebaba42b498de01b132a14cf7 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c7a3f6e45934b51a0fceeb72f0e9b91 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f18739a71d2449fb68fe5c9d62c4155 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc206a38b30d4588b4815584acebd155 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72e54c0cede2421db5c67016c5b3af51 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f67bb0c0c745a79406a241df5b8fac : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6af6ffd5c91e4fe8be6f23f4843ac438 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3b78f41c653560ac6cd3abf9f58619127b904274","author":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"date":"2019-10-24 10:55:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"56ceea0f0b4cb9a1cb33b8023823fbddeee964db","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-10-24 13:13:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"ffaa06ea37196a021914b1d769384d556f2c57a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:13:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75dd31615bc0824f22b2943413553bfa0e8cbf63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:25:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2bf401efbfde086ac804ba54f1c1356a4d096476","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:25:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/eed40818df914e42866bbaf04a832483 : SUCCESS in 3m 29s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/5c7f1eb664bf41c6a156a6879524f385 : SUCCESS in 6m 04s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/641392815f4149d6a751efdd5eb78b05 : SUCCESS in 3m 52s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/c47b2332e66b44578a7838c8f7158e75 : SUCCESS in 5m 11s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/ea8c5d40b5d047c391382abae61e0ed3 : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d72901a9075c46098504130558d644a7 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d6cde3cbc824ff0909da35feba6241c : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34c726ad9e054e9d8afa90610550aa91 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efabd5ef50c346ccb892e488d9e88c35 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/681052a52149411180aa670f6048606c : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8cf217fd22246548175c56614cd6902 : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2babceac1152441cb3dc0953958d2d9d : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c9249e47ebe6bbff9f6fa871e7b63d63d13dc825","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:27:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e3368b8fd846cc9310510f8daeafe1 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a7a5b643a7649dfacfb1e3e7442632f : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be52774f84ef56d7834a305454640f2217c527b8","revisions":{"be52774f84ef56d7834a305454640f2217c527b8":{"kind":"REWORK","_number":1,"created":"2019-10-24 10:38:48.000000000","uploader":{"_account_id":31197,"name":"Son Pham Viettel","email":"sonpn13@viettel.com.vn","username":"sonpn13"},"ref":"refs/changes/17/690917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/17/690917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/17/690917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/17/690917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/17/690917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/17/690917/1"}}},"commit":{"parents":[{"commit":"b40d7c835fc5d7646bcf03ac3c380ec9bd4e9f98","subject":"Use Horizon project template for django jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/b40d7c835fc5d7646bcf03ac3c380ec9bd4e9f98"}]}],"author":{"name":"root","email":"mrjohan3004@gmail.com","date":"2019-10-24 10:36:22.000000000","tz":420},"committer":{"name":"sonpn13","email":"sonpn13@viettel.com.vn","date":"2019-10-24 10:38:45.000000000","tz":420},"subject":"Add PDF generation support for Searchlight-UI","message":"Add PDF generation support for Searchlight-UI\n\nChange-Id: Ia259132f4c102468e15f51900c786634d5f45b95\nSigned-off-by: sonpn13 \u003csonpn13@viettel.com.vn\u003e\nStory: #2006118\nTask: #35501\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/be52774f84ef56d7834a305454640f2217c527b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/be52774f84ef56d7834a305454640f2217c527b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
