)]}'
{"id":"openstack%2Fsearchlight-ui~694700","triplet_id":"openstack%2Fsearchlight-ui~master~Icd355a3efdec419b28af5c2a277e2dc23523011d","project":"openstack/searchlight-ui","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Icd355a3efdec419b28af5c2a277e2dc23523011d","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2019-11-17 15:31:02.000000000","updated":"2019-11-21 16:17:48.000000000","submitted":"2019-11-21 16:16:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694700-1574352980859-29596582","meta_rev_id":"63474045b3253ad155e865085e719e900f741ad7","_number":694700,"virtual_id_number":694700,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"date":"2019-11-21 15:50:19.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 16:16:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2019-11-17 15:49:05.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2019-11-17 15:49:05.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-17 15:49:05.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-11-21 16:16:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21a5cd665fee8eb934d872067e2fe7bc8a663e48","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-17 15:31:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5c80ac303aa49719fb8b2d1f7b94e198fd63f3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 15:42:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/68fbe5c8a70544efbd45f032208925bd : SUCCESS in 2m 10s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/b949c224e79d4908bdbfa1b1bae5a07c : SUCCESS in 4m 46s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/c2888c8d7c0a4b5ca311a2e3c9592f14 : SUCCESS in 4m 11s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/59f9f5372e9c404693bea34a6b17a8cd : SUCCESS in 5m 23s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/1f9b9490dacf4fa78ab7363e399d2a49 : SUCCESS in 8m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec774aff5bb54100b833a737abe0bf0c : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9334f640db3c4dcdbbaa751fdadaa980 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b45acc2659904c489085d9bea40ca11b : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/addb23edd5114fa28d258b34f0c18514 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2afa53b4772f4ae18a34513728bd397f : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b7ac30e827f4858951d61cfa6a1d122 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e372f70fbec5a2ea98b76bb55bc7049fab2a6c98","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-11-17 15:49:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"1fb95a9b8d66a19a468795dbecdfb33962391b4c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-21 15:50:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"431475840ca5674ee8ed0ad38d4bee3c3328dfa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:04:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9fa57cc76891480f8274441e6ae2ffb3 : SUCCESS in 2m 30s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/44a2683e9b024eca8f2d40b825d0aedd : SUCCESS in 6m 02s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/6b0a6b817d3641538d23d1c60354beb8 : SUCCESS in 4m 04s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/fb98fb34660a47278f902a7e03fa0fdb : SUCCESS in 5m 57s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/114e1ad48c524c9cb84c91a27b16968f : SUCCESS in 6m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00fe404a4d6d43d8a83fa3f5065d4e88 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b023d53dd6748bcbbf1ac50affdf9c8 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37874a19c701496e9161aa0b6c0c5d77 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a27690a61ec745219632522b8c3e7f48 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f4b148b30e8436eb2ecb26def3c4529 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4eb07b42d6d74c14ba5443be32ff4db9 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c03e7dbf216e96ba7182b15b08c32f71cea11873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:04:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19b022612f15aa3571fbe53d9a4e332ca1284e52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:16:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"947602b182ab82592d7c1769b47293abc0e15fb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:16:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/14fdcc3e2e0c48a9a0f5b3474793824a : SUCCESS in 3m 06s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/c83fc2bf42024d71aaf9da002b02a73a : SUCCESS in 4m 51s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/3cf370a2236843beb38015d59c985641 : SUCCESS in 4m 51s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/7ea0434e21b7469a9cbf6a4fdf34c39a : SUCCESS in 5m 32s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/e0b4b05bc3da4bf9957df4670e9bd410 : SUCCESS in 6m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2105a4d7d8eb475ea7eeb31da650e2e5 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f89609928fd4a37ab28edf6a5e8b800 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c14dd1558b34b24836a95bde69f5fd5 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbc8f7bf816a4d4889b495f17956e0e0 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6243b81bcb6f496782df278cb0e004f3 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63ac17756eb1481e8ff1eef2f21668c8 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e69af857f790016b790af70cdf3daf48f9323a91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:17:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82a28921be254e30811ab9d39f7392b7 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4ad60962b0a4cc3a26174f0f51cb490 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cda5279eba34be50cf61eb8243cf4cc263db99ff","revisions":{"cda5279eba34be50cf61eb8243cf4cc263db99ff":{"kind":"REWORK","_number":1,"created":"2019-11-17 15:31:02.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/00/694700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/00/694700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/00/694700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/00/694700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/00/694700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/00/694700/1"}}},"commit":{"parents":[{"commit":"1a1ded8494299ce4e768ac8c731aa36b18fb8cc6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/1a1ded8494299ce4e768ac8c731aa36b18fb8cc6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-17 15:30:17.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-17 15:31:00.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nsearchlight-ui is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nDepends-On: https://review.opendev.org/#/c/693631/\nChange-Id: Icd355a3efdec419b28af5c2a277e2dc23523011d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/cda5279eba34be50cf61eb8243cf4cc263db99ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/cda5279eba34be50cf61eb8243cf4cc263db99ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
