)]}'
{"id":"openstack%2Fsearchlight-ui~695897","triplet_id":"openstack%2Fsearchlight-ui~master~Idfc847be0a9e9610efba0c6837b8d67a7b325c6c","project":"openstack/searchlight-ui","branch":"master","hashtags":[],"change_id":"Idfc847be0a9e9610efba0c6837b8d67a7b325c6c","subject":"Replace success with then","status":"MERGED","created":"2019-11-25 12:52:28.000000000","updated":"2019-11-25 16:38:01.000000000","submitted":"2019-11-25 16:36:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695897-1574699779117-ca8dfbed","meta_rev_id":"da28bd3583e118bed686de5e1adf540a57877d98","_number":695897,"virtual_id_number":695897,"owner":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19360,"name":"hojin kim","email":"khoj@osci.kr","username":"osc_khoj"},{"value":0,"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},{"value":0,"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 16:36:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":19360,"name":"hojin kim","email":"khoj@osci.kr","username":"osc_khoj"},{"value":0,"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},{"value":0,"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},{"value":2,"date":"2019-11-25 16:24:08.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"date":"2019-11-25 12:55:16.000000000","_account_id":19360,"name":"hojin kim","email":"khoj@osci.kr","username":"osc_khoj"},{"value":0,"date":"2019-11-25 12:54:58.000000000","_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},{"value":0,"date":"2019-11-25 12:55:32.000000000","_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},{"value":1,"date":"2019-11-25 16:24:08.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-25 12:54:33.000000000","_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19360,"name":"hojin kim","email":"khoj@osci.kr","username":"osc_khoj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},{"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 12:54:33.000000000","updated_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"reviewer":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"state":"REVIEWER"},{"updated":"2019-11-25 12:54:58.000000000","updated_by":{"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},"reviewer":{"_account_id":31300,"name":"SHINGOON","email":"shingoon7@gmail.com","username":"SHINGOON"},"state":"REVIEWER"},{"updated":"2019-11-25 12:55:16.000000000","updated_by":{"_account_id":19360,"name":"hojin kim","email":"khoj@osci.kr","username":"osc_khoj"},"reviewer":{"_account_id":19360,"name":"hojin kim","email":"khoj@osci.kr","username":"osc_khoj"},"state":"REVIEWER"},{"updated":"2019-11-25 12:55:32.000000000","updated_by":{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},"reviewer":{"_account_id":31238,"name":"Youngju Lee","email":"leeyj7141@gmail.com","username":"leeyj7141"},"state":"REVIEWER"},{"updated":"2019-11-25 16:36:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1de1fed4c7001d07dc6a07936d045c571f4ea7c5","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-11-25 12:52:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f680f9a06df958a1b4ee9e2e64f7391133a8377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 13:05:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/d3152342ce744ee7a0ff59bedad41ac1 : SUCCESS in 5m 13s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/fc4fb199517e4ea3a0722ec4aa43a111 : SUCCESS in 7m 36s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/51535e9baf1348cdaa3fce62e605a275 : SUCCESS in 8m 40s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/cbc20a3360d943289e175f239bb3d3ed : SUCCESS in 8m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6bbcea57a834bbb98cffb70b473dbf7 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb5b51deace4d25919babac7bd2ed75 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68b7854844fb48d6938483386f9833f6 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1a463df448a4020b921247be57a2de5 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242a6d6042764a8da62e4da9e1837dbe : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a5523c5e5ec72987f486c1f2ed5cdfb651391f5a","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-11-25 16:24:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd1a78e159c64b05e7477a7149907c0fb07ade9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 16:24:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94e8d6f06b8323d2aabeb4ebc00623a8b11d97d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 16:36:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/5fc8113c3778401bb6e9fb65bd146b41 : SUCCESS in 6m 40s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/727ac02b139542d2b3b372aa2ecfa2cc : SUCCESS in 6m 56s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/3664f79a818f47a5ab093ab41c565fcf : SUCCESS in 5m 32s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/55db3de81506402a8397767b54669df9 : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d821112c9704849bf78c306a8bba1b4 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/197cf0dbfeaf4d25bd9d430a405b1201 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5647eee50cca40fca0b923dbf1d9b990 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2750ce0009d646d5a6b3a42cc72c57e2 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32e5efbf56604f6d9bc4c2229f28d03f : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"187df59489436ca02fdee99dd8a07e35708e67ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 16:36:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da28bd3583e118bed686de5e1adf540a57877d98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 16:38:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66caa3cb3d94406994b5092f6f039ba4 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92bc4ab02007130ee5bd65c442137b45f5c110d2","revisions":{"92bc4ab02007130ee5bd65c442137b45f5c110d2":{"kind":"REWORK","_number":1,"created":"2019-11-25 12:52:28.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/97/695897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/97/695897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/97/695897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/97/695897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/97/695897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/97/695897/1"}}},"commit":{"parents":[{"commit":"cda5279eba34be50cf61eb8243cf4cc263db99ff","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/cda5279eba34be50cf61eb8243cf4cc263db99ff"}]}],"author":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-11-25 12:51:56.000000000","tz":540},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-11-25 12:52:24.000000000","tz":540},"subject":"Replace success with then","message":"Replace success with then\n\nChange-Id: Idfc847be0a9e9610efba0c6837b8d67a7b325c6c\nSigned-off-by: Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/92bc4ab02007130ee5bd65c442137b45f5c110d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/92bc4ab02007130ee5bd65c442137b45f5c110d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
