)]}'
{"id":"openstack%2Fsearchlight-ui~700645","triplet_id":"openstack%2Fsearchlight-ui~master~I5eb8b97475e1e277426a112f33212f78bb2295c6","project":"openstack/searchlight-ui","branch":"master","topic":"babel-config","hashtags":[],"change_id":"I5eb8b97475e1e277426a112f33212f78bb2295c6","subject":"translation: drop babel extractor definitions","status":"MERGED","created":"2019-12-26 17:27:23.000000000","updated":"2019-12-27 03:14:37.000000000","submitted":"2019-12-27 03:13:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700645-1577416403108-cc347a4a","meta_rev_id":"d2f0498b1215862f3c79ff92c90c1f3a941b971f","_number":700645,"virtual_id_number":700645,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-27 03:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2019-12-27 02:57:25.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2019-12-27 02:57:25.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-27 02:57:25.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-12-27 03:13:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d3ebd28df17e4d6c18b97865e4f8678278fdf758","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-26 17:27:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a18ad7e7358638507a580ed07222dd1b327c039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 18:03:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/e84ea3727bee43d79ea9fde67df51248 : SUCCESS in 4m 32s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/397b384e45684363adaa46a6598341a2 : SUCCESS in 4m 18s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/fa3ad7ab0b8c47e3b586252ff5211b8b : SUCCESS in 5m 32s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/cb8e2898a66a4b11b8937ff3e0cc9a8e : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92148ef8b2434c4b80c58cfb18b8b289 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c6be40061b45be95ed735abcc398f8 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/475f46de71f84a219073a4aad3f902ac : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9eb1dc659be145278a91db94a2a31964 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c124e8ecfb5d43a98e48036ae6777a09 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d0cbc92afcefd583124bf3c3d2c67683a2317a76","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-12-27 02:57:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"28a9f4c8d8b0c02b579cebb6a56d9975584369fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-27 02:57:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbd51689de5e71837c48b8eff70a864dd4f5db07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-27 03:13:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/960a14bcb41e431d83a3b5f55d9b6327 : SUCCESS in 4m 24s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/d266645812e046288a627c05df72ce62 : SUCCESS in 3m 55s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/7e2f317da6d949b9905c6425fa47285c : SUCCESS in 5m 50s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/8cdd565056994317954f90cd0beda8c3 : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2dd788734b524aebb584291d7bf3993e : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8812e880384412c8f0f97b1fd175598 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/002815da315544b4987a0f30508ad733 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b395d558637b44b88cae5426cadfaf20 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017ff670223647b9832878b2aa868247 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"de66c629aa18015680e45bcdf1f304fde3b4a486","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-27 03:13:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"659d0313280f864fe12758ccaf9fffa65ba249fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-27 03:14:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d3f1498919f4a3bac42d27fac4e8dbb : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a249a0e2cd7d4b3dad2b8310e54d7330e75d325a","revisions":{"a249a0e2cd7d4b3dad2b8310e54d7330e75d325a":{"kind":"REWORK","_number":1,"created":"2019-12-26 17:27:23.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/45/700645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/45/700645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/45/700645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/45/700645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/45/700645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/45/700645/1"}}},"commit":{"parents":[{"commit":"92bc4ab02007130ee5bd65c442137b45f5c110d2","subject":"Replace success with then","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/92bc4ab02007130ee5bd65c442137b45f5c110d2"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-26 17:27:13.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-26 17:27:13.000000000","tz":540},"subject":"translation: drop babel extractor definitions","message":"translation: drop babel extractor definitions\n\nbabel extractors are now registered via python entry points,\nso there is no need to declare babel extractors in babel configs.\n\nThis change is important to make translation work in Django 2.2.\ndjango-babel does not work with Django 2.2 and looks unmaintained\nfor over two years. The horizon team is thinking to switch the extractor\nto enmerkar (a fork of django-babel) to make extraction of translation\nstring work again near future. It is important to drop the extractor\ndefinition to make the transition smooth.\n\nChange-Id: I5eb8b97475e1e277426a112f33212f78bb2295c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/a249a0e2cd7d4b3dad2b8310e54d7330e75d325a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/a249a0e2cd7d4b3dad2b8310e54d7330e75d325a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
