)]}'
{"id":"openstack%2Fsearchlight-ui~702086","triplet_id":"openstack%2Fsearchlight-ui~master~Ic521d8283c1b5719516b8e8a15fa04a0a9ef7a04","project":"openstack/searchlight-ui","branch":"master","topic":"remove-six","hashtags":[],"change_id":"Ic521d8283c1b5719516b8e8a15fa04a0a9ef7a04","subject":"Remove six usage","status":"MERGED","created":"2020-01-11 16:01:37.000000000","updated":"2020-01-12 13:10:58.000000000","submitted":"2020-01-12 13:09:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702086-1578834578594-0548d2f5","meta_rev_id":"1b90eacde2803b308ecdd4ac024a79456f110eb4","_number":702086,"virtual_id_number":702086,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-12 13:09:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":2,"date":"2020-01-11 16:52:02.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-01-12 12:59:29.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-01-12 12:59:29.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-11 16:52:02.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-01-12 12:59:29.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-01-12 13:09:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3b0261bb13dbf77edf61acdf4fccd357ac3eba3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-11 16:01:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"367cf9934531e8b5c29ddc8946211f760e1645e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 16:12:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/50f8545a04944f369027c85a8a2d3098 : SUCCESS in 4m 49s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/bfd51ab1a16b42a6a8ee518b0fffc388 : SUCCESS in 5m 25s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/394269859a0245449de1b0d8d9feecd2 : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d90eac30cb4c46cdb43c0628e42b6fba : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adafb33543a74f77ad044e30782c8f0f : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53bdf9057b2f4c6b9d21bb115ca56053 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc43ceb2133946b3ad38d8c25a45a8bc : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e452f281b64b819c0dc8828348eac6 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0a5104c702059b029f323a02710a1c2b6927261a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-11 16:52:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44a1da5c4db06d8b91ff3f1271128c7767f0223e","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-01-12 12:59:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"f37f3541c63b9d5aee54bbbf90d97ea4c5847994","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 12:59:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90b835f138f21b7a5f486a4aef2e51663a13740c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 13:09:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d8c4ba458b9daeb89635e7c2569297e1ad0ce300","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 13:09:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/a2e550aa53ef4eddbc4043e636570217 : SUCCESS in 4m 51s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/a518a5acdfc742fea7de182d05da3aa2 : SUCCESS in 6m 03s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/7a2e0c7b23664fd7b92de894d8fd666c : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c7c7bc697f6461bb9de2e86c1c56263 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47d0a8c2006a4abe94b311806741cb9c : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c5beb821fe340acb9c6d418ce2a51cc : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3999e03bd5db4d8b919938f85a0a67e0 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55db86ebb8114005a79559237efb095a : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"eec95d32cd94f5f4bc517f573c5fb3358d06163f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 13:10:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f33d5523794d15953186364d4a95af : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa03509bcf0c75ea82edb914e49cd90186b94c79","revisions":{"aa03509bcf0c75ea82edb914e49cd90186b94c79":{"kind":"REWORK","_number":1,"created":"2020-01-11 16:01:37.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/86/702086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/86/702086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/86/702086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/86/702086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/86/702086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/86/702086/1"}}},"commit":{"parents":[{"commit":"4cf804782cd7a64abb629f3da070c54d36c7e1e9","subject":"Drop Django 1.11 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/4cf804782cd7a64abb629f3da070c54d36c7e1e9"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-11 16:01:20.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-11 16:01:20.000000000","tz":60},"subject":"Remove six usage","message":"Remove six usage\n\nThis repo does not support Python 2 anymore, so we don\u0027t need\nsix for compatibility between Python2 and 3, convert six usage to Python\n3 code.\n\nChange-Id: Ic521d8283c1b5719516b8e8a15fa04a0a9ef7a04\nNeeded-By: https://review.opendev.org/701743\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/aa03509bcf0c75ea82edb914e49cd90186b94c79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/aa03509bcf0c75ea82edb914e49cd90186b94c79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
