)]}'
{"id":"openstack%2Fsearchlight-ui~737464","triplet_id":"openstack%2Fsearchlight-ui~master~I25e67299e5f13f2036680c210bc6f0eb8e60a4cf","project":"openstack/searchlight-ui","branch":"master","topic":"fix-npm-jobs","hashtags":[],"change_id":"I25e67299e5f13f2036680c210bc6f0eb8e60a4cf","subject":"Fix npm jobs","status":"MERGED","created":"2020-06-23 09:14:31.000000000","updated":"2020-06-23 09:55:01.000000000","submitted":"2020-06-23 09:53:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737464-1592905999319-cf4bca08","meta_rev_id":"d65e22e397685bdecdd7a4d9edb45d800fd26150","_number":737464,"virtual_id_number":737464,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-23 09:53:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2020-06-23 09:41:50.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2020-06-23 09:41:50.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 09:41:50.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-06-23 09:53:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2e2262f844dee68d2c167a15ef24752f85147ef","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-06-23 09:14:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd457e68c434c6a33b9630a5cdda24ba67e220df","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-06-23 09:24:18.000000000","message":"Topic set to fix-npm-jobs","accounts_in_message":[],"_revision_number":1},{"id":"c1b3f94e2dbfece49c018c6a23cfa5deef50214a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 09:27:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4561fc8afe984178bd15edeceebf9dc2 : SUCCESS in 5m 04s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/8cf5d7218482437db66e7b89b4b4f89c : SUCCESS in 5m 22s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/f2ba6d74032641c0979e6bd7368ba526 : SUCCESS in 6m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9853e7bd96904b2a9dba715ec536fc56 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c26ceac910f7441ca1ed080c14577a7f : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b05b0af8f1bd4ef2b2c0c8db8a800274 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d542ede6a388435cb395e644e6cb4d6b : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d29ae4e4389fd55dcd96ca419b76f3bec50b9f13","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-06-23 09:41:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"8ea4e9006658d48eafe8e2c5f514e4ca6a57bdd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 09:42:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aec2514af9f5f0479e6fcf89deacc4947c36f0d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 09:53:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b1083f50dad235972616a3463b474442cb5fd9a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 09:53:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9fdd0fff553499daed440a49c02bcb0 : SUCCESS in 5m 57s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/d227d45c330642729fe1345ce41ebc14 : SUCCESS in 6m 05s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/9145345860034d2a8b6afa63717594a5 : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/545d81539f794cbab338b948893a6820 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38e2e93cf51d4504ad151faba5fa591a : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a3ba59397cc432b929e28f4a5464b6f : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a153ad84bcd4af3b575d83e8520e372 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d65e22e397685bdecdd7a4d9edb45d800fd26150","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 09:55:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/426cf1cd65434671b03762c34d46faae : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d98f5f53d05d97f42f13fc884abd9cf19ee92b38","revisions":{"d98f5f53d05d97f42f13fc884abd9cf19ee92b38":{"kind":"REWORK","_number":1,"created":"2020-06-23 09:14:31.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/64/737464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/64/737464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/64/737464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/64/737464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/64/737464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/64/737464/1"}}},"commit":{"parents":[{"commit":"13383c25c9b13e9972de43bd1e65c41187f08100","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/13383c25c9b13e9972de43bd1e65c41187f08100"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-06-23 09:13:45.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-06-23 09:14:23.000000000","tz":0},"subject":"Fix npm jobs","message":"Fix npm jobs\n\nChange-Id: I25e67299e5f13f2036680c210bc6f0eb8e60a4cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/d98f5f53d05d97f42f13fc884abd9cf19ee92b38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/d98f5f53d05d97f42f13fc884abd9cf19ee92b38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
