)]}'
{"id":"openstack%2Fsearchlight-ui~744585","triplet_id":"openstack%2Fsearchlight-ui~master~I0e0c8e9cf604a75ba3cec8c4d14dad479974f44c","project":"openstack/searchlight-ui","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I0e0c8e9cf604a75ba3cec8c4d14dad479974f44c","subject":"migrate testing to ubuntu focal","status":"MERGED","created":"2020-08-04 01:55:01.000000000","updated":"2020-08-08 13:39:57.000000000","submitted":"2020-08-08 13:38:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744585-1596893919433-453d58e8","meta_rev_id":"bb7da73785f6be9469f2238ed525999252a9006f","_number":744585,"virtual_id_number":744585,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-08 13:38:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2020-08-04 13:10:38.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-08-07 11:51:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-08 13:26:43.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-08 13:26:43.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 13:10:38.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-08-07 11:51:29.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2020-08-08 13:26:43.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-08-08 13:38:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f82bee739ae47b5e9412402e925d4251e1ed32b1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 01:55:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa74a2480431533be3681df6ed31bc0e1eea4a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 02:13:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/764b805a19034bf2ba30bbbee3a49646 : SUCCESS in 4m 39s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/f681dc618a564c2d961ec9959ad30b09 : SUCCESS in 4m 07s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/5a83393c928f467ab8175f02c5b77af4 : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f30539e6b6334e789ad5ac0a4edfc8a8 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8477dafcf31848188d13ea2674e7b936 : FAILURE in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d529e5cb6db4542a1422140a67c1469 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29f0d3e3eb8b40d0b7e0604fa99b3384 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8b3d4a4708ec360a8f015c7cfd3d517271740c0a","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-04 10:54:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"224fd113c646dc65f3a904b9128fa63bddc42f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 11:05:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d7f9933c432c4b1a93c7e6b6fe774d53 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bdb9955822742f3860d1d5584e41072 : SUCCESS in 4m 24s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/0d72816ecdeb4e0f8009feb5db4e4ff2 : SUCCESS in 4m 41s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/ee850cf3ee064c68a62d6cdb7492d7f5 : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c20173548cec4cdd81e6f1000675e8e1 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c35b299b83b646c681b1ae9442274bb4 : FAILURE in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/562198ba4e9f41f583a0741d5342b7bb : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/956c70e0b63b43dd9bea50dd15a68a4e : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c289d476edfd489ab51b4e7917129e36 : SUCCESS in 4m 26s\n\nWarning:\n  Comments left for invalid file searchlight_ui/tests/api_tests/test_searchlight_rest.py","accounts_in_message":[],"_revision_number":2},{"id":"40ba022c1bd1cff454c46ab717503c269af85974","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-04 12:27:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8a23949fd369e67fbd7cd0befd27c4fa6b347e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 12:38:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c29533ca8154c718d023c3f8e22b479 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4aa1e8484a1843f7a99d3caa4d04b8a7 : SUCCESS in 5m 02s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/7d7de5b24b7b4719a4ed572a2407db63 : SUCCESS in 5m 23s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/658c20586054410cb56c6daa49e06d79 : SUCCESS in 6m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d34027a3ea44be6818e092b304497ce : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78dd376419e44eb181136284e9cb43c8 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04766e711d0b4966929fcf7f080a2f13 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c665514a737e4d93a85558c6da49231c : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1caa4506b0e041bdaada4f1f0660e66d : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"c0002ff26082287dee00b34ce77e97af1b4adcb9","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-04 13:05:28.000000000","message":"Patch Set 3: Code-Review+1\n\nall working here now, let\u0027s remove the depends-on now to get this merge before base jobs switch to Focal and update commit message.","accounts_in_message":[],"_revision_number":3},{"id":"d6586fdb687a74db95b7a095b2576becceadac6c","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-04 13:10:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a8e446253862729380b5ea226d9e9988532d22cf","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-04 13:10:38.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"89c5c4b18f93a8adce496f0a03f03dcc80445da7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 13:23:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/03869f12fd2f428781b7b209ff444e61 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ffce0dbbde94eda88c9283b39b60536 : SUCCESS in 5m 13s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/54f9ad90733e4830ac3e046651117afe : SUCCESS in 5m 40s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/709c608234274589aa39f921a7f6288b : SUCCESS in 7m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35824dd62fcc486b81e9eb5e80db5b50 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0077202c25624d78aed541cb7395cf98 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfdc0a022c934ac4a032c8ba13f64ad6 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d49c74196184320b7ee16510b149806 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46e4f5bd952947f3b7e4731a036a955d : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":4},{"id":"04e169b17d5d30f70feea5cf18852982972d468f","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2020-08-07 11:51:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"495525b3f0fa8f5e473361c2120158d745851e57","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-08-08 13:26:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":4},{"id":"f1b378228829f15b437fefaaf11c51934ca17b1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 13:26:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"681fc4382632e9195362743e5e412384628b10c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 13:38:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7c1fd1dc5f9d4973a4187c5da4439d39 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/535d4395cd394dc2a129ec5286d87dab : SUCCESS in 3m 59s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/3f276b9fb49f4f3593ebaa122088b363 : SUCCESS in 4m 18s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/831e3e802a6a4d45bf413f87b0382592 : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c993aed973042afa39bdd0100789af5 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04ba33fa7981424790b9c0cb53dbbfb4 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc90798918854e3b84508b76bd76408d : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26ba02d51104540ad571f81832850ce : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98b8a5fecb314100a6c322b33dad27b8 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":4},{"id":"b62f2228f77ff71c0dbec50bd5c699db64e24539","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 13:38:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"0fb604be54215d1cac3f27b2b5db56fd038c2ec0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 13:39:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40731a3f66a743e98db4ef06a53d2df0 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9e31509fedc4842908e784b4b946e7e : SUCCESS in 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7aa5bbf556c1e25eb3e252b04a6865a3f9afd45d","revisions":{"b36692fae94a6e3b47287b9ae1fb4a2daf27d08f":{"kind":"REWORK","_number":1,"created":"2020-08-04 01:55:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/85/744585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/85/744585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/1"}}},"commit":{"parents":[{"commit":"eea664cad33169a723dc99f5ff1de31679e1bf2e","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/eea664cad33169a723dc99f5ff1de31679e1bf2e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 01:52:42.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 01:55:00.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40217\n\nDepends-On: https://review.opendev.org/#/c/734700/\nChange-Id: I0e0c8e9cf604a75ba3cec8c4d14dad479974f44c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/b36692fae94a6e3b47287b9ae1fb4a2daf27d08f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/b36692fae94a6e3b47287b9ae1fb4a2daf27d08f"}]},"branch":"refs/heads/master"},"9410640bf83ab39fa9030e39ea874e9e2eb34b61":{"kind":"REWORK","_number":2,"created":"2020-08-04 10:54:08.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/85/744585/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/85/744585/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/2"}}},"commit":{"parents":[{"commit":"eea664cad33169a723dc99f5ff1de31679e1bf2e","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/eea664cad33169a723dc99f5ff1de31679e1bf2e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 01:52:42.000000000","tz":-300},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-04 10:53:23.000000000","tz":0},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40217\n\nDepends-On: https://review.opendev.org/#/c/734700/\nChange-Id: I0e0c8e9cf604a75ba3cec8c4d14dad479974f44c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/9410640bf83ab39fa9030e39ea874e9e2eb34b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/9410640bf83ab39fa9030e39ea874e9e2eb34b61"}]},"branch":"refs/heads/master"},"7849d090bede8f8ecc972b7f1f0e30390c108c73":{"kind":"REWORK","_number":3,"created":"2020-08-04 12:27:18.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/85/744585/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/85/744585/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/3"}}},"commit":{"parents":[{"commit":"eea664cad33169a723dc99f5ff1de31679e1bf2e","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/eea664cad33169a723dc99f5ff1de31679e1bf2e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 01:52:42.000000000","tz":-300},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-04 12:26:34.000000000","tz":0},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40217\n\nDepends-On: https://review.opendev.org/#/c/734700/\nChange-Id: I0e0c8e9cf604a75ba3cec8c4d14dad479974f44c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/7849d090bede8f8ecc972b7f1f0e30390c108c73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/7849d090bede8f8ecc972b7f1f0e30390c108c73"}]},"branch":"refs/heads/master"},"7aa5bbf556c1e25eb3e252b04a6865a3f9afd45d":{"kind":"REWORK","_number":4,"created":"2020-08-04 13:10:06.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/85/744585/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/85/744585/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/85/744585/4"}}},"commit":{"parents":[{"commit":"eea664cad33169a723dc99f5ff1de31679e1bf2e","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/eea664cad33169a723dc99f5ff1de31679e1bf2e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 01:52:42.000000000","tz":-300},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-04 13:07:31.000000000","tz":0},"subject":"migrate testing to ubuntu focal","message":"migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1886296\nBump the pyflakes to 2.1.1 as min version to run pep8 jobs\non py3.8 which is default python vesion in ubuntu focal.\n\nStory: #2007865\nTask: #40217\n\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\nChange-Id: I0e0c8e9cf604a75ba3cec8c4d14dad479974f44c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/7aa5bbf556c1e25eb3e252b04a6865a3f9afd45d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/7aa5bbf556c1e25eb3e252b04a6865a3f9afd45d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
