)]}'
{"id":"openstack%2Fsearchlight-ui~747263","triplet_id":"openstack%2Fsearchlight-ui~master~I6f6d55ab6728430d95f23fabc5edf5df80af2c5e","project":"openstack/searchlight-ui","branch":"master","hashtags":[],"change_id":"I6f6d55ab6728430d95f23fabc5edf5df80af2c5e","subject":"[DNM] Test searchlight-ui with horizon jobs on focal","status":"ABANDONED","created":"2020-08-20 18:14:04.000000000","updated":"2020-08-21 11:21:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e8eec42892b9579beab2b5bdef85c0aeb7d0cb53","_number":747263,"virtual_id_number":747263,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-20 18:38:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 18:38:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9644dfca3911a299a542b5dd63a136ba408fca3","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-20 18:14:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d662cb7b43f93a38b09170d4912c5e464a1702ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 18:38:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a09ad7ea61c462ba11410e04ba7595a : SUCCESS in 5m 06s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/dbc110d4ce41436baf298bf5e7291c4e : SUCCESS in 5m 47s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/53aaaf16d3994d7186ce2cb17a9576a5 : SUCCESS in 9m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5af273a42a04ce99f7dbd37429d0ff5 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0073b72a6a9e4bf48913362edb924b70 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a0d87c213f447ca8553057a5687f78c : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dda8b1faa3254266a4b9bb08b50036ba : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e8eec42892b9579beab2b5bdef85c0aeb7d0cb53","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-21 11:21:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1931554f771d739c756ab63b3f559462d4634ca7","revisions":{"1931554f771d739c756ab63b3f559462d4634ca7":{"kind":"REWORK","_number":1,"created":"2020-08-20 18:14:04.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/63/747263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/63/747263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/63/747263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/63/747263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/63/747263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/63/747263/1"}}},"commit":{"parents":[{"commit":"7aa5bbf556c1e25eb3e252b04a6865a3f9afd45d","subject":"migrate testing to ubuntu focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/7aa5bbf556c1e25eb3e252b04a6865a3f9afd45d"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-20 18:12:49.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-20 18:12:49.000000000","tz":0},"subject":"[DNM] Test searchlight-ui with horizon jobs on focal","message":"[DNM] Test searchlight-ui with horizon jobs on focal\n\nDepends-On: https://review.opendev.org/#/c/744647/\nChange-Id: I6f6d55ab6728430d95f23fabc5edf5df80af2c5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/1931554f771d739c756ab63b3f559462d4634ca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/1931554f771d739c756ab63b3f559462d4634ca7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
