)]}'
{"id":"openstack%2Fsearchlight-ui~756320","triplet_id":"openstack%2Fsearchlight-ui~master~I5f3aaa009c16e7d2339251a36a2845a893a55773","project":"openstack/searchlight-ui","branch":"master","hashtags":[],"change_id":"I5f3aaa009c16e7d2339251a36a2845a893a55773","subject":"Dep\u0027s should be restricted by upper-constraints","status":"ABANDONED","created":"2020-10-06 14:24:06.000000000","updated":"2020-12-07 18:38:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2ed6803b29b9f8a564fd2b9bf91bc34c76cbe23e","_number":756320,"virtual_id_number":756320,"owner":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-06 14:56:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2020-11-19 08:45:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":-1,"date":"2020-11-19 14:53:04.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-06 14:29:33.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 14:29:33.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-10-06 14:56:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 08:45:09.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"df619d27baf12cb5f968788f1238f83bc3363e3d","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-10-06 14:24:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8d0a8b401d3f731d3507dd3274035abc2fddf28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 14:56:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7666c117be8f4f7a8434b6014e74ce5f : SUCCESS in 6m 29s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/95e49c9ee0104666aab5684df347dcb0 : SUCCESS in 4m 37s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/1fb332f331314cfe8ed1e3d6731e432d : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e798596a1d824123ac3eb77878fa7ad2 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc28885bfd9f43caae7e16742f0cb83d : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e13605efd524cb2a781c430362bf7e0 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f4c333be79464c84a676137beb2061 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bde35c99e324d8fae078d8e7b9fc48e : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"798af2e8e17505a9f1cea2ae171371b169e61531","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-11-19 08:45:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"915f41b37621f05a9e99a0c85fa16e947b579a47","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-11-19 14:53:04.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"2ed6803b29b9f8a564fd2b9bf91bc34c76cbe23e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-07 18:38:43.000000000","message":"Abandoned\n\nProject Retirement: http://lists.openstack.org/pipermail/openstack-discuss/2020-November/018637.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8a24bfe9744e788a3f814d72faa51578e694357","revisions":{"b8a24bfe9744e788a3f814d72faa51578e694357":{"kind":"REWORK","_number":1,"created":"2020-10-06 14:24:06.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/20/756320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight-ui","ref":"refs/changes/20/756320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/20/756320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/20/756320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight-ui refs/changes/20/756320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight-ui refs/changes/20/756320/1"}}},"commit":{"parents":[{"commit":"ee7c731486fc2bfc592dd74b5081108c624c2f72","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/ee7c731486fc2bfc592dd74b5081108c624c2f72"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-10-06 14:23:52.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-10-06 14:24:04.000000000","tz":480},"subject":"Dep\u0027s should be restricted by upper-constraints","message":"Dep\u0027s should be restricted by upper-constraints\n\nTox trying to install latest versions for building docs which may\nnot be supported by stable and lower branches, so should be\nrestricted by respective version\u0027s upper-constraints.txt\n\nChange-Id: I5f3aaa009c16e7d2339251a36a2845a893a55773\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/b8a24bfe9744e788a3f814d72faa51578e694357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight-ui/commit/b8a24bfe9744e788a3f814d72faa51578e694357"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
