)]}'
{"id":"openstack%2Fsearchlight~361813","triplet_id":"openstack%2Fsearchlight~master~I37b1d59a7c93b8fa55e65b9b21ca681d3db96551","project":"openstack/searchlight","branch":"master","topic":"bug/remove-logging-unused","hashtags":[],"change_id":"I37b1d59a7c93b8fa55e65b9b21ca681d3db96551","subject":"TrivialFix: Remove logging import unused","status":"MERGED","created":"2016-08-29 03:12:27.000000000","updated":"2016-08-30 15:03:02.000000000","submitted":"2016-08-30 15:03:02.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"73617569d8ac1fc5789d3a8baa1f3ea87015ded7","_number":361813,"virtual_id_number":361813,"owner":{"_account_id":15905,"name":"Cao Xuan Hoang","email":"xuanhoangqb@gmail.com","username":"hoangcx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},{"value":2,"date":"2016-08-30 15:03:02.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4428,"name":"Liyingjun","email":"yinjalee@163.com","username":"liyingjun"},{"value":0,"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},"all":[{"value":2,"date":"2016-08-30 14:56:13.000000000","_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-08-30 07:12:50.000000000","_account_id":4428,"name":"Liyingjun","email":"yinjalee@163.com","username":"liyingjun"},{"value":1,"date":"2016-08-30 07:13:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},"all":[{"value":1,"date":"2016-08-30 14:56:13.000000000","_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4428,"name":"Liyingjun","email":"yinjalee@163.com","username":"liyingjun"},{"value":0,"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4428,"name":"Liyingjun","email":"yinjalee@163.com","username":"liyingjun"},{"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-08-30 07:12:50.000000000","updated_by":{"_account_id":4428,"name":"Liyingjun","email":"yinjalee@163.com","username":"liyingjun"},"reviewer":{"_account_id":4428,"name":"Liyingjun","email":"yinjalee@163.com","username":"liyingjun"},"state":"REVIEWER"},{"updated":"2016-08-30 07:13:28.000000000","updated_by":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"reviewer":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"state":"REVIEWER"},{"updated":"2016-08-30 14:56:13.000000000","updated_by":{"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},"reviewer":{"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},"state":"REVIEWER"},{"updated":"2016-08-30 15:03:02.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"e17b2fd7d4ffca7fbb41968005965c87fbb5e654","author":{"_account_id":15905,"name":"Cao Xuan Hoang","email":"xuanhoangqb@gmail.com","username":"hoangcx"},"date":"2016-08-29 03:12:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d62249dc451b664ac4f8d393486330d2f24d8e4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-29 03:17:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-searchlight-docs-ubuntu-xenial http://docs-draft.openstack.org/13/361813/1/check/gate-searchlight-docs-ubuntu-xenial/17366d9//doc/build/html/ : SUCCESS in 2m 24s\n- gate-searchlight-pep8-ubuntu-xenial http://logs.openstack.org/13/361813/1/check/gate-searchlight-pep8-ubuntu-xenial/4019a99/ : SUCCESS in 4m 25s\n- gate-searchlight-python27-searchlight http://logs.openstack.org/13/361813/1/check/gate-searchlight-python27-searchlight/694d199/ : SUCCESS in 4m 38s\n- gate-searchlight-python27-searchlight-es1 http://logs.openstack.org/13/361813/1/check/gate-searchlight-python27-searchlight-es1/d359c84/ : FAILURE in 3m 22s (non-voting)\n- gate-searchlight-python34-searchlight http://logs.openstack.org/13/361813/1/check/gate-searchlight-python34-searchlight/a625844/ : SUCCESS in 3m 41s\n- gate-searchlight-python35-searchlight-nv http://logs.openstack.org/13/361813/1/check/gate-searchlight-python35-searchlight-nv/7b39f45/ : SUCCESS in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecb3ab890e4407402bbee18fa729a8a06ff5de44","author":{"_account_id":4428,"name":"Liyingjun","email":"yinjalee@163.com","username":"liyingjun"},"date":"2016-08-30 07:12:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"11b8713764c3ee88f63b243ae95821babc288611","author":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"date":"2016-08-30 07:13:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9007fe5b40078d1ddfcaebd3c4a5d1f34f10ac4e","author":{"_account_id":7665,"name":"Travis Tripp","email":"os.travis.tripp@gmail.com","username":"travis.tripp"},"date":"2016-08-30 14:56:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"807ecc01b73db937f3ea7a004d1f2840e80a7723","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-30 14:56:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"18c3f87eb333912ea896c9effd2369695c17bbac","date":"2016-08-30 15:03:02.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"620073d706a4259dc5812865c0573dad3e1177be","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-30 15:03:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-searchlight-docs-ubuntu-xenial http://docs-draft.openstack.org/13/361813/1/gate/gate-searchlight-docs-ubuntu-xenial/db7f7a7//doc/build/html/ : SUCCESS in 3m 20s\n- gate-searchlight-pep8-ubuntu-xenial http://logs.openstack.org/13/361813/1/gate/gate-searchlight-pep8-ubuntu-xenial/cdb115a/ : SUCCESS in 3m 09s\n- gate-searchlight-python27-searchlight http://logs.openstack.org/13/361813/1/gate/gate-searchlight-python27-searchlight/e9a5597/ : SUCCESS in 4m 30s\n- gate-searchlight-python27-searchlight-es1 http://logs.openstack.org/13/361813/1/gate/gate-searchlight-python27-searchlight-es1/6831d3b/ : FAILURE in 4m 18s (non-voting)\n- gate-searchlight-python34-searchlight http://logs.openstack.org/13/361813/1/gate/gate-searchlight-python34-searchlight/112c702/ : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d783c877513ae2650cebe042d8a784cc9e0b36bb","revisions":{"d783c877513ae2650cebe042d8a784cc9e0b36bb":{"kind":"REWORK","_number":1,"created":"2016-08-29 03:12:27.000000000","uploader":{"_account_id":15905,"name":"Cao Xuan Hoang","email":"xuanhoangqb@gmail.com","username":"hoangcx"},"ref":"refs/changes/13/361813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/13/361813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/13/361813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/13/361813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/13/361813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/13/361813/1"}}},"commit":{"parents":[{"commit":"458f2cb215bdacc0ece793c46a6f4c17994463eb","subject":"Merge \"Get ready for os-api-ref sphinx theme change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/458f2cb215bdacc0ece793c46a6f4c17994463eb"}]}],"author":{"name":"Cao Xuan Hoang","email":"hoangcx@vn.fujitsu.com","date":"2016-08-29 03:09:17.000000000","tz":420},"committer":{"name":"Cao Xuan Hoang","email":"hoangcx@vn.fujitsu.com","date":"2016-08-29 03:09:51.000000000","tz":420},"subject":"TrivialFix: Remove logging import unused","message":"TrivialFix: Remove logging import unused\n\nThis patch removes logging import unused in\nsearchlight/notifier.py\n\nChange-Id: I37b1d59a7c93b8fa55e65b9b21ca681d3db96551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/d783c877513ae2650cebe042d8a784cc9e0b36bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/d783c877513ae2650cebe042d8a784cc9e0b36bb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"458f2cb215bdacc0ece793c46a6f4c17994463eb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
