)]}'
{"id":"openstack%2Fsearchlight~653967","triplet_id":"openstack%2Fsearchlight~master~I9be3ebed89b59230fc281dfd17cba4c2d6d3237c","project":"openstack/searchlight","branch":"master","hashtags":[],"change_id":"I9be3ebed89b59230fc281dfd17cba4c2d6d3237c","subject":"Migrate to opendev.org","status":"MERGED","created":"2019-04-20 07:38:29.000000000","updated":"2019-04-22 02:21:33.000000000","submitted":"2019-04-22 02:21:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653967-1555899693347-dfe5b5ea","meta_rev_id":"b3ac40a7e92da8de3ebf516e83a1056ccf5d0111","_number":653967,"virtual_id_number":653967,"owner":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-04-22 02:21:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-22 02:11:48.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-22 02:11:48.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 02:21:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4be6fed390c6a4674c018f901e5136882e3cedaa","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-04-20 07:38:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f2fe7d2473153784846d9edc41d160ebe591794","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 07:47:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/67/653967/1/check/test-release-openstack/00917a8/ : SUCCESS in 2m 52s\n- openstack-tox-pep8 http://logs.openstack.org/67/653967/1/check/openstack-tox-pep8/2cf5ba9/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/67/653967/1/check/openstack-tox-py27/3e739a4/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/67/653967/1/check/openstack-tox-py36/4952ea1/ : SUCCESS in 6m 25s\n- openstack-tox-py37 http://logs.openstack.org/67/653967/1/check/openstack-tox-py37/517ea63/ : SUCCESS in 4m 48s\n- openstack-tox-cover http://logs.openstack.org/67/653967/1/check/openstack-tox-cover/8f52920/cover/ : SUCCESS in 6m 50s\n- openstack-tox-docs http://logs.openstack.org/67/653967/1/check/openstack-tox-docs/3717e23/html/ : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/653967/1/check/openstack-tox-lower-constraints/9a605e4/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2b274bd0f1afe3e945d8f680a1abb80fc59f8db6","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-04-20 07:50:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a19acb78192241519ef4c25c3972c0bfa683eea6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 08:01:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/67/653967/2/check/test-release-openstack/7c99ff5/ : SUCCESS in 2m 05s\n- openstack-tox-pep8 http://logs.openstack.org/67/653967/2/check/openstack-tox-pep8/a1279c0/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/67/653967/2/check/openstack-tox-py27/837c81c/ : FAILURE in 6m 58s\n- openstack-tox-py36 http://logs.openstack.org/67/653967/2/check/openstack-tox-py36/2d5aff0/ : SUCCESS in 6m 38s\n- openstack-tox-py37 http://logs.openstack.org/67/653967/2/check/openstack-tox-py37/e10d44f/ : FAILURE in 7m 11s\n- openstack-tox-cover http://logs.openstack.org/67/653967/2/check/openstack-tox-cover/ff8d4c1/ : FAILURE in 7m 22s\n- openstack-tox-docs http://logs.openstack.org/67/653967/2/check/openstack-tox-docs/4293e7a/html/ : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/653967/2/check/openstack-tox-lower-constraints/1d080a2/ : SUCCESS in 7m 21s\n- build-openstack-releasenotes http://logs.openstack.org/67/653967/2/check/build-openstack-releasenotes/eec2a16/html/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"0288e7cf563626b046eb330c286707f193038225","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-04-20 08:12:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6885d872304b22e9a1b1c2c9602a802a182cb754","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 08:18:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/67/653967/2/check/test-release-openstack/353ee46/ : SUCCESS in 2m 30s\n- openstack-tox-pep8 http://logs.openstack.org/67/653967/2/check/openstack-tox-pep8/eb9bc65/ : SUCCESS in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/67/653967/2/check/openstack-tox-py27/a95185a/ : SUCCESS in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/67/653967/2/check/openstack-tox-py36/c249634/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/67/653967/2/check/openstack-tox-py37/3f5dcec/ : SUCCESS in 5m 16s\n- openstack-tox-cover http://logs.openstack.org/67/653967/2/check/openstack-tox-cover/fe24cb2/cover/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/67/653967/2/check/openstack-tox-docs/805b05e/html/ : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/653967/2/check/openstack-tox-lower-constraints/633dce8/ : SUCCESS in 4m 47s\n- build-openstack-releasenotes http://logs.openstack.org/67/653967/2/check/build-openstack-releasenotes/c1a811f/html/ : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"205f84bc3df9a1b1277d0213a144eda2fb720a9c","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-04-22 01:43:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"76174a05104bff7c7e5543ed0d9a089f1f468f5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 01:54:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/67/653967/3/check/test-release-openstack/ca49607/ : SUCCESS in 2m 19s\n- openstack-tox-pep8 http://logs.openstack.org/67/653967/3/check/openstack-tox-pep8/9a6083f/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/67/653967/3/check/openstack-tox-py27/32463d5/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/67/653967/3/check/openstack-tox-py36/b484aa8/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/67/653967/3/check/openstack-tox-py37/ffd2cb5/ : SUCCESS in 5m 04s\n- openstack-tox-cover http://logs.openstack.org/67/653967/3/check/openstack-tox-cover/54c8a36/cover/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/67/653967/3/check/openstack-tox-docs/cc72e9a/html/ : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/653967/3/check/openstack-tox-lower-constraints/586f160/ : SUCCESS in 4m 31s\n- build-openstack-releasenotes http://logs.openstack.org/67/653967/3/check/build-openstack-releasenotes/ebf59f8/html/ : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":3},{"id":"6e14e2291d7e83e6331d61224708f4ee78b1d112","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-04-22 02:00:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"be3e3e32e35de8c5ecea569895b2726797a6abf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 02:10:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/67/653967/4/check/test-release-openstack/2ef7d08/ : SUCCESS in 2m 31s\n- openstack-tox-pep8 http://logs.openstack.org/67/653967/4/check/openstack-tox-pep8/fcec20e/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/67/653967/4/check/openstack-tox-py27/1b9133e/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/67/653967/4/check/openstack-tox-py36/b8c4c9c/ : SUCCESS in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/67/653967/4/check/openstack-tox-py37/7c250f6/ : SUCCESS in 5m 00s\n- openstack-tox-cover http://logs.openstack.org/67/653967/4/check/openstack-tox-cover/4752445/cover/ : SUCCESS in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/67/653967/4/check/openstack-tox-docs/25bd3fd/html/ : SUCCESS in 3m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/653967/4/check/openstack-tox-lower-constraints/e073985/ : SUCCESS in 5m 24s\n- build-openstack-releasenotes http://logs.openstack.org/67/653967/4/check/build-openstack-releasenotes/f64882c/html/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":4},{"id":"9d71308ba9e78ec771e0216508351094d4bff9d5","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-04-22 02:11:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6827f74d4cffb3f9668f6a6336a3d74c3467ea04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 02:12:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"94fc1486287da053f1fb22a3b94ecd55f974ca78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 02:21:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b3ac40a7e92da8de3ebf516e83a1056ccf5d0111","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 02:21:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/67/653967/4/gate/test-release-openstack/4a9d6d3/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/67/653967/4/gate/openstack-tox-pep8/f729673/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/67/653967/4/gate/openstack-tox-py27/507b371/ : SUCCESS in 5m 12s\n- openstack-tox-py36 http://logs.openstack.org/67/653967/4/gate/openstack-tox-py36/6ad3f84/ : SUCCESS in 6m 54s\n- openstack-tox-py37 http://logs.openstack.org/67/653967/4/gate/openstack-tox-py37/2be4d9c/ : SUCCESS in 7m 04s\n- openstack-tox-docs http://logs.openstack.org/67/653967/4/gate/openstack-tox-docs/a22c59d/html/ : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/653967/4/gate/openstack-tox-lower-constraints/855f3b1/ : SUCCESS in 4m 39s\n- build-openstack-releasenotes http://logs.openstack.org/67/653967/4/gate/build-openstack-releasenotes/85c5df5/html/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3374cba06c651f4be196ad2fe8f6435fd20dfd35","revisions":{"7dbbb1a7808a201d0891d58e939da861e93f3b7b":{"kind":"REWORK","_number":1,"created":"2019-04-20 07:38:29.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/67/653967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/67/653967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/67/653967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/67/653967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/67/653967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/67/653967/1"}}},"commit":{"parents":[{"commit":"f56ef4262df709007ed005a400d49cf2d86b0de7","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/f56ef4262df709007ed005a400d49cf2d86b0de7"}]}],"author":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-04-20 07:38:07.000000000","tz":540},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-04-20 07:38:07.000000000","tz":540},"subject":"Migrate to opendev.org","message":"Migrate to opendev.org\n\nChange-Id: I9be3ebed89b59230fc281dfd17cba4c2d6d3237c\nSigned-off-by: Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/7dbbb1a7808a201d0891d58e939da861e93f3b7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/7dbbb1a7808a201d0891d58e939da861e93f3b7b"}]},"branch":"refs/heads/master"},"ce0c8f276201de825edd809a6698ea3a9f1dd81a":{"kind":"REWORK","_number":2,"created":"2019-04-20 07:50:17.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/67/653967/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/67/653967/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/67/653967/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/67/653967/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/67/653967/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/67/653967/2"}}},"commit":{"parents":[{"commit":"f56ef4262df709007ed005a400d49cf2d86b0de7","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/f56ef4262df709007ed005a400d49cf2d86b0de7"}]}],"author":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-04-20 07:38:07.000000000","tz":540},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-04-20 07:50:08.000000000","tz":540},"subject":"Migrate to opendev.org","message":"Migrate to opendev.org\n\nChange-Id: I9be3ebed89b59230fc281dfd17cba4c2d6d3237c\nSigned-off-by: Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/ce0c8f276201de825edd809a6698ea3a9f1dd81a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/ce0c8f276201de825edd809a6698ea3a9f1dd81a"}]},"branch":"refs/heads/master"},"060097558fa5b89354b345d2f78142baa146eb36":{"kind":"REWORK","_number":3,"created":"2019-04-22 01:43:50.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/67/653967/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/67/653967/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/67/653967/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/67/653967/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/67/653967/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/67/653967/3"}}},"commit":{"parents":[{"commit":"f56ef4262df709007ed005a400d49cf2d86b0de7","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/f56ef4262df709007ed005a400d49cf2d86b0de7"}]}],"author":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-04-20 07:38:07.000000000","tz":540},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-04-22 01:43:41.000000000","tz":540},"subject":"Migrate to opendev.org","message":"Migrate to opendev.org\n\nChange-Id: I9be3ebed89b59230fc281dfd17cba4c2d6d3237c\nSigned-off-by: Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/060097558fa5b89354b345d2f78142baa146eb36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/060097558fa5b89354b345d2f78142baa146eb36"}]},"branch":"refs/heads/master"},"3374cba06c651f4be196ad2fe8f6435fd20dfd35":{"kind":"REWORK","_number":4,"created":"2019-04-22 02:00:44.000000000","uploader":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"ref":"refs/changes/67/653967/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/searchlight","ref":"refs/changes/67/653967/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/67/653967/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/67/653967/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/searchlight refs/changes/67/653967/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/searchlight refs/changes/67/653967/4"}}},"commit":{"parents":[{"commit":"f56ef4262df709007ed005a400d49cf2d86b0de7","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/f56ef4262df709007ed005a400d49cf2d86b0de7"}]}],"author":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-04-20 07:38:07.000000000","tz":540},"committer":{"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","date":"2019-04-22 02:00:35.000000000","tz":540},"subject":"Migrate to opendev.org","message":"Migrate to opendev.org\n\nChange-Id: I9be3ebed89b59230fc281dfd17cba4c2d6d3237c\nSigned-off-by: Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/3374cba06c651f4be196ad2fe8f6435fd20dfd35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/searchlight/commit/3374cba06c651f4be196ad2fe8f6435fd20dfd35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
